)]}'
{"id":"openstack%2Fpuppet-swift~712864","triplet_id":"openstack%2Fpuppet-swift~master~Ia5112083b8e9d2ebc7e933191c4c8e51d2e48903","project":"openstack/puppet-swift","branch":"master","topic":"centos-8-jobs","hashtags":[],"change_id":"Ia5112083b8e9d2ebc7e933191c4c8e51d2e48903","subject":"Use centos-8 standalone puppet parent","status":"MERGED","created":"2020-03-13 08:50:33.000000000","updated":"2020-03-13 21:39:26.000000000","submitted":"2020-03-13 21:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"712864-1584135566245-962e1dba","meta_rev_id":"c0b167a74e6be533c5febb4db375a2f445db1937","_number":712864,"virtual_id_number":712864,"owner":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 21:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-03-13 12:18:11.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-03-13 20:12:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-03-13 20:12:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-13 12:18:11.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-03-13 20:12:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-03-13 21:39:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16a3181cc9ea8618aeaf4f1572f299aec77613cf","author":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"date":"2020-03-13 08:50:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"95152506d65d02095f75780283cb950ab147fbcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 10:26:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd5272079cf1471db7d3b4495115e3de : SUCCESS in 5m 31s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f8991c7fc20489387efbc0cd3cf0e54 : SUCCESS in 5m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23501887a5704280b502a7ea32047880 : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0946e528e3534959b2422346956752ed : SUCCESS in 25m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/078b2d924fce4303be8b4c1b555ebd8e : SUCCESS in 18m 24s (non-voting)\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/d4470490dc8c4d28aba301776cdf201e : SUCCESS in 18m 34s (non-voting)\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4eafa75cfe8249ebace45ba1e8ae822c : SUCCESS in 17m 45s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/daf2d7c515444b5eaaab02636958300a : SUCCESS in 15m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4597fe76b62e4a37969a0216b5e46138 : SUCCESS in 49m 02s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/58041bee06264fbebf8abe7d4e332866 : SUCCESS in 1h 04m 43s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bb764874ed4d4198b9a0bf03441b57cb : SUCCESS in 1h 00m 33s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1a8635278c2345f38b99e956f6f86b01 : SUCCESS in 25m 40s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abd3f71f30574fc5ba8df6e4c19e7b25 : FAILURE in 23m 00s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5b4058c8d47b41659a7caf87807db4b9 : SUCCESS in 1h 31m 45s","accounts_in_message":[],"_revision_number":1},{"id":"0e46c9ae62890abc9eede5d2230db94bad8f5d1a","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-03-13 12:18:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f9449267fc56818934f476ba840ffb61c2e9f321","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-03-13 20:12:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6648124e00bde49386f732f69532cc127b75a7a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 20:13:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"298b227836e31aad378f3b2a98a97ee6e6409218","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:39:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5ca27755752d72672696c743952b8b09b5ffc6af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 21:39:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29bc834ac11b418d9dc1799db14d0e24 : SUCCESS in 5m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c033384e2dd455ca21dde2b4dd868ad : SUCCESS in 5m 09s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04c14f4892fb4e36b5e92e02f85d0943 : SUCCESS in 5m 36s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6660b570fd7948029fe3064e5c977331 : SUCCESS in 17m 52s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f70071d4d4841b39c2cbf075b3be4f1 : SUCCESS in 13m 29s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3fbce826c07640c6949e13f01add186b : SUCCESS in 1h 00m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/48b8f0841ba241658bfc1ec3576a2c5a : SUCCESS in 1h 10m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ad9ef8cc73844aec9c4277473ffccb40 : SUCCESS in 29m 53s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/34e348b3176b4d838e08b76f3c8c7eaf : SUCCESS in 1h 24m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aaf7edc758533982be569489a5d083d15516c922","revisions":{"aaf7edc758533982be569489a5d083d15516c922":{"kind":"REWORK","_number":1,"created":"2020-03-13 08:50:33.000000000","uploader":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"ref":"refs/changes/64/712864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/64/712864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/64/712864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/64/712864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/64/712864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/64/712864/1"}}},"commit":{"parents":[{"commit":"de7eb6fff1d908e329d2bbd539e645dfc8680b65","subject":"Merge \"Deprecate swift::proxy::ceilometer::auth_uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de7eb6fff1d908e329d2bbd539e645dfc8680b65"}]}],"author":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-13 08:49:01.000000000","tz":330},"committer":{"name":"Chandan Kumar (raukadah)","email":"chkumar@redhat.com","date":"2020-03-13 08:50:18.000000000","tz":330},"subject":"Use centos-8 standalone puppet parent","message":"Use centos-8 standalone puppet parent\n\nhttps://review.opendev.org/#/c/711320/ adds\ntripleo-puppet-ci-centos-8-standalone job in poi and let\u0027s\nre-use the same parent for tripleo standalone job.\n\nChange-Id: Ia5112083b8e9d2ebc7e933191c4c8e51d2e48903\nSigned-off-by: Chandan Kumar (raukadah) \u003cchkumar@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/aaf7edc758533982be569489a5d083d15516c922"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/aaf7edc758533982be569489a5d083d15516c922"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
