)]}'
{"id":"openstack%2Fpuppet-swift~721557","triplet_id":"openstack%2Fpuppet-swift~master~I3cd3c76210da75271459b8079c23604e828f10d9","project":"openstack/puppet-swift","branch":"master","topic":"custom-uuid-fact","hashtags":[],"change_id":"I3cd3c76210da75271459b8079c23604e828f10d9","subject":"Guess devices UUIDs","status":"ABANDONED","created":"2020-04-21 09:15:50.000000000","updated":"2022-05-04 17:40:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"04d3120e9a8e9338b8e215c5f1bffa33fd677f6d","_number":721557,"virtual_id_number":721557,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-21 11:18:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-21 11:18:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2dd64a1d6bc28821614728b4272af7d503a3d056","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-04-21 09:15:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"36b839b754a0c982e66fa04be75c8831e88e8eb4","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-04-21 09:37:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d9cc79131f96e3995142472c19bfd9ce937beed7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 11:18:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/328c9a4b1c5f4cb19c47a2b12dfa76b6 : SUCCESS in 5m 00s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a245bf0c7e4478e9a8cad9fb039f3a0 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9058879429f94f06872ce98bf9a1296b : SUCCESS in 4m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7cd6c7456ca74111a06e60a3c3125a4b : FAILURE in 19m 24s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/9586d153a5a24b2890eb1a699d907722 : FAILURE in 18m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e3954f57f7c48138ee74a96ef9d6d83 : FAILURE in 15m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f2c968cafe344dab116e5ec2fc530de : FAILURE in 11m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b45527ad850546bca4e419ecdf642b87 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/148ad6fb143f46ac862028b19cc3e123 : SUCCESS in 1h 05m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2bdb9b24fb5746f0bcd0707f8fe20af5 : FAILURE in 10m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f5182d27a2d2480cbc24f1b727211bfe : FAILURE in 8m 53s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5fc09a04c46f4cd4906f5063cc4145ff : FAILURE in 7m 13s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/21408ea1b9954d8384c4475f61ae1365 : SUCCESS in 1h 36m 11s","accounts_in_message":[],"_revision_number":2},{"id":"04d3120e9a8e9338b8e215c5f1bffa33fd677f6d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-04 17:40:42.000000000","message":"Abandoned\n\nI\u0027ll abandon this change because it has been inactive for 1+ year. Please feel free to reopen this if you resume your work.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef39586e13bd51f70b9d43a4c9bb9af0b5c13454","revisions":{"c340a8bcfae9c0be0dc00e29885c882e3b9805c5":{"kind":"REWORK","_number":1,"created":"2020-04-21 09:15:50.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/57/721557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/57/721557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/1"}}},"commit":{"parents":[{"commit":"d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-04-21 09:14:51.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-04-21 09:14:51.000000000","tz":120},"subject":"Guess devices UUIDs","message":"Guess devices UUIDs\n\nThis patch adds a custom Ruby fact to guess device UUIDs\nusing blkid on the storage machine.\n\nChange-Id: I3cd3c76210da75271459b8079c23604e828f10d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c340a8bcfae9c0be0dc00e29885c882e3b9805c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c340a8bcfae9c0be0dc00e29885c882e3b9805c5"}]},"branch":"refs/heads/master"},"ef39586e13bd51f70b9d43a4c9bb9af0b5c13454":{"kind":"REWORK","_number":2,"created":"2020-04-21 09:37:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/57/721557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/57/721557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/57/721557/2"}}},"commit":{"parents":[{"commit":"d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-04-21 09:14:51.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-04-21 09:37:45.000000000","tz":120},"subject":"Guess devices UUIDs","message":"Guess devices UUIDs\n\nThis patch adds a custom Ruby fact to guess device UUIDs\nusing blkid on the storage machine.\n\nChange-Id: I3cd3c76210da75271459b8079c23604e828f10d9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef39586e13bd51f70b9d43a4c9bb9af0b5c13454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef39586e13bd51f70b9d43a4c9bb9af0b5c13454"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
