)]}'
{"id":"openstack%2Fpuppet-swift~722559","triplet_id":"openstack%2Fpuppet-swift~master~Id28f47840f46ecdfe463eb718a1fcb823b7ca109","project":"openstack/puppet-swift","branch":"master","topic":"etag-quoter","hashtags":[],"change_id":"Id28f47840f46ecdfe463eb718a1fcb823b7ca109","subject":"Add etag_quoter middleware support","status":"MERGED","created":"2020-04-24 02:05:46.000000000","updated":"2020-04-27 15:20:14.000000000","submitted":"2020-04-27 15:14:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722559-1588000463899-bde3a662","meta_rev_id":"3dd747f9e643d9cc30763604d16f307b97e41405","_number":722559,"virtual_id_number":722559,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 15:14:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-04-24 21:29:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2020-04-26 14:52:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-04-27 13:14:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 14:52:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-27 13:14:24.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-27 15:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5001705a3b8f7fada2e9a74bfcd2bd545892655","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-24 02:05:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b87e11e3945616e41c1a554808dad2ce53afd280","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-24 02:13:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7f447f5d673b1e2ffd6d0caed2bf44ce0863b2cd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-24 02:14:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8da736f331995f39ad1f8c92772722285398afe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 03:40:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/70c3c485b55c4022ad02f12feb7a4b6e : FAILURE in 4m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94446804a1134adb9dc74c805510e84d : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da44db4d42b6458e9cbbd7c7d6a6b5c5 : SUCCESS in 4m 11s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/3995fe8fff5e49c981ba75b4466f0f60 : SUCCESS in 14m 32s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/58b057c4215546628cdd9bfc58eda5f4 : SUCCESS in 15m 22s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef4fbdca7c27465d87403ae096a65f1d : SUCCESS in 13m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2058a3dd711b4fde881cce7c7aa04193 : SUCCESS in 14m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d4c417510f08400a94973338d4a969ed : SUCCESS in 47m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7fe04609cd7448149a17ac1d9603f943 : SUCCESS in 1h 01m 29s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/831bb22f34b249e2985e351b57ebe5dc : SUCCESS in 25m 55s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/05bdb311e7db43db9ff98d61ca7af753 : FAILURE in 7m 00s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c0f8ce6bcb544a1aea7aacdcb0c9307 : FAILURE in 6m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85361aed701c4595b7958758800f46d5 : SUCCESS in 3m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4170d938180f4783b690fbb233b560db : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":2},{"id":"5b3ff7fc2eb61c7e54f4fb8a9b91ec24a2eacebc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-24 07:38:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"93fe2233d3aadab416d9ccbc5e84cfe5c34ef14c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 10:39:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47b2ba372b1f4121ac2ab3563d7375d3 : SUCCESS in 5m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4607563ea363437497a9836623e0bfbb : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e437846c7ff34348b3ae7d9fcd39d98d : SUCCESS in 5m 21s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d61bf651008d4f518882ae6e225eeaa8 : SUCCESS in 17m 59s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/3b2c6c7af7a647ac9a615cc1e0ec9bf9 : SUCCESS in 18m 36s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/393d46b6bf62457da3646e0e4fa42711 : SUCCESS in 16m 18s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1c8ec77f3fbb4c3abc254ebc7c7eee80 : SUCCESS in 18m 08s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9ee1cfd3371497bbd5f025cc5ffce75 : SUCCESS in 47m 15s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a2688f8dea814d00b89cba5bee2e786a : SUCCESS in 1h 04m 59s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/dfeb329bd6674a2b9503d45f6804d721 : SUCCESS in 28m 37s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/48127bca687145feaff19c6cc5687d4c : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48bad4443c274474a998fed0d361091c : FAILURE in 11m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2f1cf50c0a44ec9b71b54a8672c1004 : SUCCESS in 4m 12s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a769905a288445bfb5836d426beacc2d : SUCCESS in 1h 31m 57s","accounts_in_message":[],"_revision_number":3},{"id":"030492cde06fe6fbdde8e043f46c8250348bb50a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-24 21:29:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a1ef925c1ff9fb9002e062d6e3eb19c5ea089d56","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-26 14:52:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9658c4763be41ece204c9968a676009f23d44c4f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-27 13:14:24.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3381ae22428173675d0e1dee077bca6307fcd2b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:28:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce5df3765b346df49fba8e4fd5c1cbdf4d20d9ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:14:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58c2263d01cf47de8eb16c1cd0fb4401 : SUCCESS in 6m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d352867f8323493c8c62af53886af21c : SUCCESS in 8m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b347ba2a0a6d4a328b7b5fa1caf3aca0 : SUCCESS in 7m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/713b781e0909497088c29d408a13e511 : SUCCESS in 19m 29s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/3f39f62c3d564541a66af099c8c89579 : SUCCESS in 20m 07s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8f4ff61e8e04d8ba76d27dba8c7480d : SUCCESS in 18m 21s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/960ae598e7d34c9c8a52e8d3202b7937 : SUCCESS in 54m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5de13eafab4748d488dcbe9c795b7546 : SUCCESS in 1h 14m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2e97842964f94f96bb5eb81551c5e8b6 : SUCCESS in 32m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2297e839356f4c78b7a68b7af30048b8 : SUCCESS in 6m 07s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/34d76ba98d754ea8a4f8ca34153d7e6d : SUCCESS in 1h 40m 18s","accounts_in_message":[],"_revision_number":3},{"id":"8f088fbdb8d55cc9ed9e8769bf86d18090aa7b68","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:14:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d28b85abbfbe26abb5885171af9682693589c4c5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:20:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a33ef32f1f634fa9aa4ac8f292f54a5d : SUCCESS in 1m 44s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"38ebeea216d78ca06be59122055007373362b5bb","revisions":{"8fa2b6e607cf364186ec1a23679e27d211527626":{"kind":"REWORK","_number":1,"created":"2020-04-24 02:05:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/722559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/59/722559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/1"}}},"commit":{"parents":[{"commit":"d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 02:05:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 02:05:11.000000000","tz":540},"subject":"Add etag_quoter middleware support","message":"Add etag_quoter middleware support\n\nChange-Id: Id28f47840f46ecdfe463eb718a1fcb823b7ca109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8fa2b6e607cf364186ec1a23679e27d211527626"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8fa2b6e607cf364186ec1a23679e27d211527626"}]},"branch":"refs/heads/master"},"ad1004bfff5b4ba01a12c8c0d116c79076cef984":{"kind":"REWORK","_number":2,"created":"2020-04-24 02:14:26.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/722559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/59/722559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/2"}}},"commit":{"parents":[{"commit":"d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 02:05:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 02:14:18.000000000","tz":540},"subject":"Add etag_quoter middleware support","message":"Add etag_quoter middleware support\n\nChange-Id: Id28f47840f46ecdfe463eb718a1fcb823b7ca109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ad1004bfff5b4ba01a12c8c0d116c79076cef984"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ad1004bfff5b4ba01a12c8c0d116c79076cef984"}]},"branch":"refs/heads/master"},"38ebeea216d78ca06be59122055007373362b5bb":{"kind":"REWORK","_number":3,"created":"2020-04-24 07:38:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/59/722559/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/59/722559/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/59/722559/3"}}},"commit":{"parents":[{"commit":"d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e","subject":"Prepare Ussuri M3 again","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d9f55bdbd2f01ffb2b3656bfbfcab16bf155de8e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 02:05:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-24 07:38:44.000000000","tz":540},"subject":"Add etag_quoter middleware support","message":"Add etag_quoter middleware support\n\nChange-Id: Id28f47840f46ecdfe463eb718a1fcb823b7ca109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/38ebeea216d78ca06be59122055007373362b5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/38ebeea216d78ca06be59122055007373362b5bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
