)]}'
{"id":"openstack%2Fpuppet-swift~723224","triplet_id":"openstack%2Fpuppet-swift~stable%2Ftrain~I252584bdcef64e9d0056b132cb57bfd6f87e187b","project":"openstack/puppet-swift","branch":"stable/train","topic":"middleware","hashtags":[],"change_id":"I252584bdcef64e9d0056b132cb57bfd6f87e187b","subject":"Add symlink middleware support","status":"MERGED","created":"2020-04-26 13:14:55.000000000","updated":"2020-04-27 17:34:32.000000000","submitted":"2020-04-27 17:33:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723224-1588008784697-92884968","meta_rev_id":"a32909af816e395f9fca00f4bc5bd633e1117c33","_number":723224,"virtual_id_number":723224,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 17:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2020-04-26 16:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2020-04-27 15:00:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2020-04-27 15:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-26 16:12:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-27 15:00:18.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-04-27 17:33:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3296a3ba4b1cfb7f7878bcfb7b1be01eca257159","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-26 13:14:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ffe460c56e3a527b503a9f1404f5665794f13177","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-26 14:08:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"9e9cc5fab1b24c3dac22c01745fdabbc187f0849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 16:10:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cc398690424748668557987b16f650e6 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e229f23310b4cb08b93b73c2b91c001 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b9bef26920f44d2864a627739d618b4 : SUCCESS in 4m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/02b61701c1bb498db47bf250a48927e0 : SUCCESS in 26m 57s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bc38fa61914c4d5baff4c564d369a504 : SUCCESS in 16m 32s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/1ee91d26e1c4489f92495542761169c6 : SUCCESS in 15m 25s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7a596544b86428fbd0ff07faaa88bf1 : SUCCESS in 14m 05s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/684cce5370e74c4b8902c394e00f7591 : SUCCESS in 14m 39s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e5e9930632e4707a3480d5b2b8a09bf : SUCCESS in 43m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4beb1fdce13a4a95bc44edfb32151739 : SUCCESS in 1h 01m 35s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/dc46616c0b93417a8ce88baeed8eeb39 : SUCCESS in 57m 58s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ae6d45078e8746a8bf1962ae0215a5d4 : SUCCESS in 26m 45s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae68cebc0d71498580277a22019b7332 : FAILURE in 25m 22s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/12b422b2e2e94cbfb57b5f51949bd1b6 : SUCCESS in 1h 52m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/182a719bb3c74f65885f0e2cf71c8fd1 : SUCCESS in 3m 21s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/428808b70445424086c32fb54de41ba7 : SUCCESS in 1h 37m 16s","accounts_in_message":[],"_revision_number":2},{"id":"473633a644b762bdb1478d7e85e24a589b3fccc0","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-26 16:12:17.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"91a2ea1e940d782ac119f9a7f5f4a8d03f3dd6e1","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-04-27 15:00:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e6fa8b2eaaa6be578b07728f7340267cba94e38","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:07:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"70737f5195d77146eb82ef0ea0164ec807edec73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:33:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74e58d793b4b4d4e8c260db167279a95 : SUCCESS in 5m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5f9ba2eec6c4e059ebbd7211124ac93 : SUCCESS in 4m 59s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0be296ee421d4bb5ab45a8909bc18b5e : SUCCESS in 7m 34s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/6c9569ef86924acb9a20d5c9240e75ca : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ec422d32c2c408490084075451ee5f1 : SUCCESS in 16m 07s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3a2a7a4ad1d047248c46b6ce3b02717e : SUCCESS in 1h 07m 33s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/6f24ebbaba4b484683625db2c15b21d8 : SUCCESS in 1h 19m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f980a526273a45d8b637ca9d6c0ff98b : SUCCESS in 31m 40s\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/af18ab0382be4426941f41fb5561fd98 : SUCCESS in 2h 23m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67e6930e9ac845969f78810c9f65f8df : SUCCESS in 6m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9dda11a8995b4e3e88b9ba67ee1c0f55 : SUCCESS in 1h 36m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d5a5799c67a48b53b9310686b8b3f620d05d75d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:33:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fbb740e055d6cb34523b74e5fef84f1053a42ff3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:34:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/048b8cda3c2944328f91fea1770679ac : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1795fc0bd04d5cb70582518bfb027c0957bc525d","revisions":{"a08cfe1e1943cdd892e74d94c34eb6c5c0a1c7ef":{"kind":"REWORK","_number":1,"created":"2020-04-26 13:14:55.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/24/723224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/24/723224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/1"}}},"commit":{"parents":[{"commit":"654e96445302cd7c77376c2b5e5ed67d464f29e8","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/654e96445302cd7c77376c2b5e5ed67d464f29e8"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-23 05:51:54.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-26 13:13:25.000000000","tz":600},"subject":"Add symlink middleware support","message":"Add symlink middleware support\n\nChange-Id: I252584bdcef64e9d0056b132cb57bfd6f87e187b\n(cherry picked from commit 05ccc857d8a209f85ffab37a4b504a56d6c7d1f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a08cfe1e1943cdd892e74d94c34eb6c5c0a1c7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a08cfe1e1943cdd892e74d94c34eb6c5c0a1c7ef"}]},"branch":"refs/heads/stable/train"},"1795fc0bd04d5cb70582518bfb027c0957bc525d":{"kind":"REWORK","_number":2,"created":"2020-04-26 14:08:33.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/24/723224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/24/723224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/24/723224/2"}}},"commit":{"parents":[{"commit":"654e96445302cd7c77376c2b5e5ed67d464f29e8","subject":"Switch to Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/654e96445302cd7c77376c2b5e5ed67d464f29e8"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-23 05:51:54.000000000","tz":600},"committer":{"name":"Shi Yan","email":"yanshi.403@gmail.com","date":"2020-04-26 14:08:30.000000000","tz":0},"subject":"Add symlink middleware support","message":"Add symlink middleware support\n\nChange-Id: I252584bdcef64e9d0056b132cb57bfd6f87e187b\n(cherry picked from commit 05ccc857d8a209f85ffab37a4b504a56d6c7d1f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
