)]}'
{"id":"openstack%2Fpuppet-swift~723816","triplet_id":"openstack%2Fpuppet-swift~master~If85df5dd8fd70329500db858dc81346e52640b47","project":"openstack/puppet-swift","branch":"master","topic":"swift3","hashtags":[],"change_id":"If85df5dd8fd70329500db858dc81346e52640b47","subject":"Remove unused swift3 middleware","status":"ABANDONED","created":"2020-04-28 07:01:06.000000000","updated":"2021-01-23 07:13:29.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"47314efb511a3dc0d990f0e7a69feb153ac3441d","_number":723816,"virtual_id_number":723816,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-28 12:11:29.000000000","_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"date":"2020-04-28 12:34:29.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-28 08:56:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":-1,"date":"2020-04-28 07:51:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 07:51:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-28 08:56:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-04-28 12:34:29.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1668b6097c0b0d5734d1d52dd5b8fec342054e78","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 07:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6b56e197e32389371628cb64e479f5fbd45343f8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-28 07:10:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nJust a one nit comment regarding example manifests.\n\nIn general that we deprecate code before we actually remove it, but in this case swift3 has been deprecated 1 year ago and I don\u0027t think anybody still tries to use swift3, so I\u0027m ok to go ahead with just removing the code.\n\n[1] https://opendev.org/x/swift3/commit/211def7d590fdc7546b1a10716b145f544483fd4","accounts_in_message":[],"_revision_number":1},{"id":"e4d82d4f117c73e15b02e9607f5cc08df6e2dd2c","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 07:13:46.000000000","message":"Patch Set 1:\n\nThanks, good point, will replace it with s3api","accounts_in_message":[],"_revision_number":1},{"id":"ad6954a7dd2f0ac0f4f456b979a150bd5560c764","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 07:14:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4a0fb6054d9dc02d4dc6e5b26e0905bf5faf0fcd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-28 07:51:45.000000000","message":"Patch Set 2: Code-Review-1\n\nNeed to deprecate behavior, cannot delete.","accounts_in_message":[],"_revision_number":2},{"id":"e1ea2241178db730825e84a09f547ef050713f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 08:56:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88fd75ee9976472093449474cd7d0388 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb4c5270f0f14adf9e9c90b09a9ac354 : SUCCESS in 6m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf557a989692416fa00bd2af564d3e41 : SUCCESS in 6m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cfcfb45562f749c2b7e5a077d86fe042 : SUCCESS in 16m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d2d44fcc07794114801ffa5b949cdb79 : SUCCESS in 19m 29s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f969c5fa2d449e0bd79c56b6ad687d0 : SUCCESS in 19m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0b6c4ca81b8646afb4ff8682a552cf70 : SUCCESS in 19m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40cac0ed1929489397b16d63f4a39a3a : SUCCESS in 50m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/60dd293185bc45ce87a564ee5a7a6f41 : SUCCESS in 1h 13m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7901c305e4424131bdd2011525d69335 : SUCCESS in 29m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/680ba9cccfb24b7c85dc4b181efd9205 : FAILURE in 12m 42s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bffd1ad32048420bb5c528e2a7e98ff8 : FAILURE in 6m 06s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be40450a9a0447a7ad0a8ef5ae373e39 : SUCCESS in 7m 49s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b2fb5e77c4804217b0d6dcee6ba4f5f6 : EXCEPTION","accounts_in_message":[],"_revision_number":2},{"id":"dc4a0ab5b44ed7cc09c544585a8a13055b48ac13","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 11:15:43.000000000","message":"Patch Set 2:\n\n\u003e Need to deprecate behavior, cannot delete.\n\nI understand your workflow wanting to deprecate and then delete. But this part of the code has been changed for a long time.  So if you use this swift3 class, no matter you using stein or train, it will break as the swift3 package will be in conflict dependency for example for Debian. That is why i think we should remove it directly.","accounts_in_message":[],"_revision_number":2},{"id":"a953e699806b3d71e36102c163bd1068abfe9360","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-28 11:28:08.000000000","message":"Patch Set 2:\n\nChange so the puppet interface is not broken, fix the actual behavior in the code.","accounts_in_message":[],"_revision_number":2},{"id":"814ef6056aeb6d9399cd16a35e9c1c1cc8956e2d","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 12:11:29.000000000","message":"Patch Set 2:\n\n\u003e Change so the puppet interface is not broken, fix the actual\n \u003e behavior in the code.\n\nI don\u0027t fully understand here. I think the fix is the user not use swift3 class anymore. As a user, if I use swift3 class, it will have to install swift3 package(swift-plugin-3 for example) which is dependent on python-swift but it is wrong since we already go for python3 after stein. So we cannot make it backward compatible","accounts_in_message":[],"_revision_number":2},{"id":"1f99809d885088e818b5837f8b6b03a8e6f5e6ac","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-28 12:34:29.000000000","message":"Patch Set 2:\n\nI\u0027ve submitted an example change , which keeps current puppet class but removes some codes which doesn\u0027t work correctly. May I ask your opinion about that, Tobias?\n\nhttps://review.opendev.org/#/c/723883/","accounts_in_message":[],"_revision_number":2},{"id":"47314efb511a3dc0d990f0e7a69feb153ac3441d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 07:13:29.000000000","message":"Abandoned\n\nRemoval has been done by the different patch https://review.opendev.org/c/openstack/puppet-swift/+/758773","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e4a40b67eb5cd511affcef5da5b46300c51351e","revisions":{"eeefd514db2ab91586697e92a10e9b3dc5c0dc99":{"kind":"REWORK","_number":1,"created":"2020-04-28 07:01:06.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/16/723816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/723816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/1"}}},"commit":{"parents":[{"commit":"26b821dabbcbfb434d241eb9b733f187612aef62","subject":"Merge \"Add missing openstacklib to dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/26b821dabbcbfb434d241eb9b733f187612aef62"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 06:59:28.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 07:00:28.000000000","tz":600},"subject":"Remove unused swift3 middleware","message":"Remove unused swift3 middleware\n\nChange-Id: If85df5dd8fd70329500db858dc81346e52640b47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eeefd514db2ab91586697e92a10e9b3dc5c0dc99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eeefd514db2ab91586697e92a10e9b3dc5c0dc99"}]},"branch":"refs/heads/master"},"8e4a40b67eb5cd511affcef5da5b46300c51351e":{"kind":"REWORK","_number":2,"created":"2020-04-28 07:14:24.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/16/723816/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/16/723816/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/16/723816/2"}}},"commit":{"parents":[{"commit":"26b821dabbcbfb434d241eb9b733f187612aef62","subject":"Merge \"Add missing openstacklib to dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/26b821dabbcbfb434d241eb9b733f187612aef62"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 06:59:28.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 07:12:30.000000000","tz":600},"subject":"Remove unused swift3 middleware","message":"Remove unused swift3 middleware\n\nChange-Id: If85df5dd8fd70329500db858dc81346e52640b47\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8e4a40b67eb5cd511affcef5da5b46300c51351e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8e4a40b67eb5cd511affcef5da5b46300c51351e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
