)]}'
{"id":"openstack%2Fpuppet-swift~723818","triplet_id":"openstack%2Fpuppet-swift~stable%2Ftrain~If85df5dd8fd70329500db858dc81346e52640b47","project":"openstack/puppet-swift","branch":"stable/train","topic":"swift3","hashtags":[],"change_id":"If85df5dd8fd70329500db858dc81346e52640b47","subject":"Remove unused swift3 middleware","status":"ABANDONED","created":"2020-04-28 07:09:26.000000000","updated":"2021-01-23 07:15:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"da383d357c49d7b5e7daf402b8c9d92c9d5eb156","_number":723818,"virtual_id_number":723818,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-28 11:17:39.000000000","_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-04-28 09:00:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":-2,"date":"2020-04-28 07:51:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 07:51:28.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-04-28 09:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"305f57a382f90a12312c72a1de110901c86a0ce7","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 07:09:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7b2046a01195996bd69991f1918c42026a03774","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 07:16:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc4611139900c00c299f3f2445eadc372664a35f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-28 07:51:28.000000000","message":"Patch Set 2: Code-Review-2\n\nCannot remove code in a stable release","accounts_in_message":[],"_revision_number":2},{"id":"ff8a22ad0c0d54a8cf590e369d1a3b1e899a05b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 09:00:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/901cbfcbc0b64341b22e7a0c0192f99e : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9ddc9bda85f491aac624386450fac27 : SUCCESS in 4m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8e47814ab644e70ad1e7bb15c7858d9 : SUCCESS in 5m 26s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/9082dd9df6e048c1bbfab796f106a140 : SUCCESS in 20m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e4671c1069340339cf0ea54b53c7bc3 : SUCCESS in 16m 23s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/622af9bfed3b47579011905daf279755 : SUCCESS in 15m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b8f9f53cf834179a9019102678f7c85 : SUCCESS in 16m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8bf8d6dd64e741b68c35abd6764d2424 : SUCCESS in 19m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f4be097bc0647ff8b1cd1f0913b09b4 : SUCCESS in 57m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0bc66a4eeba74096b35ba17566e0739a : SUCCESS in 1h 02m 58s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c23aaf563c404189b57407bcd91961f8 : SUCCESS in 1h 10m 17s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e8d4c6d84f02449396018eae7a5c7d6c : SUCCESS in 27m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49c02fd0979b43b38e7a0e09a12b3cbe : FAILURE in 31m 03s (non-voting)\n- tripleo-puppet-ci-centos-7-undercloud-containers https://zuul.opendev.org/t/openstack/build/f5c6cf603f85462c817321feb59b347e : EXCEPTION\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b2720ae730b44dc836c3ce449a7b679 : SUCCESS in 4m 42s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/541a8720547341bdaaaf50230a3ac025 : SUCCESS in 1h 29m 31s","accounts_in_message":[],"_revision_number":2},{"id":"2365b2cf82e32a7e7a4157dd11e840c63b6fbfdb","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 11:17:39.000000000","message":"Patch Set 2:\n\n\u003e Cannot remove code in a stable release\n\nAs I commented in another change, since it is the breakage, I think it\u0027s worth to delete it in a stable release, just as a bug fix.","accounts_in_message":[],"_revision_number":2},{"id":"c8a03684a667106396f701377654ed11f8d6beb9","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-04-28 11:28:20.000000000","message":"Patch Set 2:\n\nChange so the puppet interface is not broken, fix the actual behavior in the code.","accounts_in_message":[],"_revision_number":2},{"id":"da383d357c49d7b5e7daf402b8c9d92c9d5eb156","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-23 07:15:18.000000000","message":"Abandoned\n\nThe different patch was merged and swift3 support has been deprecated and removed. We shouldn\u0027t backport removal to stable branches to keep interfaces so I\u0027ll abandon this.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"869d6980b8db08914ecf2c87bbe2ae47b1d24a26","revisions":{"025a2a1128d4331165745ad5632676aaaf27d552":{"kind":"REWORK","_number":1,"created":"2020-04-28 07:09:26.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/18/723818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/18/723818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/1"}}},"commit":{"parents":[{"commit":"1795fc0bd04d5cb70582518bfb027c0957bc525d","subject":"Add symlink middleware support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 06:59:28.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 07:08:25.000000000","tz":600},"subject":"Remove unused swift3 middleware","message":"Remove unused swift3 middleware\n\nChange-Id: If85df5dd8fd70329500db858dc81346e52640b47\n(cherry picked from commit eeefd514db2ab91586697e92a10e9b3dc5c0dc99)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/025a2a1128d4331165745ad5632676aaaf27d552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/025a2a1128d4331165745ad5632676aaaf27d552"}]},"branch":"refs/heads/stable/train"},"869d6980b8db08914ecf2c87bbe2ae47b1d24a26":{"kind":"REWORK","_number":2,"created":"2020-04-28 07:16:19.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/18/723818/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/18/723818/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/18/723818/2"}}},"commit":{"parents":[{"commit":"1795fc0bd04d5cb70582518bfb027c0957bc525d","subject":"Add symlink middleware support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 06:59:28.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 07:15:57.000000000","tz":600},"subject":"Remove unused swift3 middleware","message":"Remove unused swift3 middleware\n\nChange-Id: If85df5dd8fd70329500db858dc81346e52640b47\n(cherry picked from commit 8e4a40b67eb5cd511affcef5da5b46300c51351e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/869d6980b8db08914ecf2c87bbe2ae47b1d24a26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/869d6980b8db08914ecf2c87bbe2ae47b1d24a26"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
