)]}'
{"id":"openstack%2Fpuppet-swift~724034","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~I3c0344b6a202607e3d3b1380c9296ea113f5f670","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I3c0344b6a202607e3d3b1380c9296ea113f5f670","subject":"Update .gitreview for stable/ussuri","status":"MERGED","created":"2020-04-28 14:59:19.000000000","updated":"2020-05-09 08:12:04.000000000","submitted":"2020-05-09 08:12:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724034-1589011924054-d1a6a7d1","meta_rev_id":"c88c521804892c69515d0e68a0675f277fe3639e","_number":724034,"virtual_id_number":724034,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-09 08:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-04-30 09:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-09 06:38:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-09 06:38:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 09:43:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-09 06:38:39.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-09 08:12:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f843df880dc4501577d3c177975d87b7c81e005","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-04-28 14:59:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bb19960952e65f40f4592dbda1f31a8865fa9cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:34:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ad9ead6a368c4457ac8a4c6a989039d0 : SUCCESS in 7m 18s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae38328114ef45aba18368a13faaffbf : SUCCESS in 7m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/39905b91ec8f4b5a9837b3954c4fc36d : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9f5a49b87bf74d6a8b815b8e594eadaf : SUCCESS in 22m 49s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/baaf2999d6e544b1a1a3138f71b02875 : SUCCESS in 23m 47s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8035de59dd64cb6abdd6ea6f61aab55 : SUCCESS in 19m 34s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0f863b02af24983b5fd7e4a0ef1ca3d : SUCCESS in 21m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50ed25c1aa6a48a0a7f64e54508171e1 : SUCCESS in 1h 02m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d07a10f92a4c4b1594bf88b47e2a1d34 : SUCCESS in 1h 03m 32s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ffe7c7a7c834430e87bc9f7186aa74e0 : SUCCESS in 34m 39s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/62d226692a9d439b9d1b66e2c1a1755f : FAILURE in 11m 44s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bdaaa762de84231838756ba3a1608e2 : FAILURE in 10m 26s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b2548a46c0284c2bb9177a0b95a81ab9 : FAILURE in 19m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e507f0df2036495d8c4e573df1dcfc018052235a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 03:02:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea45d54e744fa04d230ba1fc122499fa27457864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 04:06:27.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2284c5f8d45f494dac4d6d205c18fe66 : SUCCESS in 4m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e21d82b18f084b9baba128ae47af22bb : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bc29e51fcfc424ea714969830e93a19 : SUCCESS in 4m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/df474dbb0f2147c68cabed9c140f5b59 : SUCCESS in 15m 39s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8204a388b8b94b88813c07504773da62 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b4134132acc409288e5441b49c8d4c6 : SUCCESS in 13m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e366b8eef6274108a2744138dbe7ab0c : SUCCESS in 16m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7971b24d32bf42de839954bbc3cfe55b : SUCCESS in 52m 24s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/39f009b9739349ec9931e7048b338894 : SUCCESS in 1h 02m 06s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/e246bd7f45d542a296c3a6cac5fa9323 : SUCCESS in 28m 32s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/047d4bd46cf14e8ab2e398df65fe117b : FAILURE in 8m 23s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21d5141afdab49cca61d222156397022 : FAILURE in 7m 59s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4f28707306de45a28bb38fe9f02dac01 : FAILURE in 13m 00s","accounts_in_message":[],"_revision_number":1},{"id":"9153d5ac0feffcc49c3202587fd5469ac7c8d9ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 15:13:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5f3aaa5577e14d044e3b82a984b3b4778b4f63c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:13:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/51a1721ba8944421896feb4f252225e9 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1c2ec052a764c73b18c0d8af0acff5c : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f97367466914e328542e715a373584d : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/8950f00a6910437e8a09744c74061afa : SUCCESS in 16m 13s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/ab64cbf5d277464aa12baf30b2aa501e : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/284184c9fa694bc2971e4259710911f4 : SUCCESS in 14m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/944c9935d2534469877ddf8243abba2a : SUCCESS in 16m 42s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09a02cc5198e49269939a2615beeed92 : SUCCESS in 51m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/689a5e2e154c411e8afb31e5304fadf5 : FAILURE in 37m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/96c9b9540dfd4ef3a0aa7031e171609e : SUCCESS in 28m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/17a4d99554a940839d938fea204b4984 : FAILURE in 10m 23s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb8d27ed3ebe485e9979c823d9e9601a : FAILURE in 7m 28s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93fe14c568aa403b9e2e0bde763bc9bd : FAILURE in 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a802355c5a34b6c5104c6fee316082fad55e4fef","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 09:43:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c37af57e98573d6fd110e96069b7980856481070","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 09:43:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e87ccad8f6df432ac86ecdbce5ffa91178d0e915","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:25:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/22ac70a66f284d949b8b01c5f8d6aece : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee05030ad6e54f43bc80ac74f4013e6a : SUCCESS in 4m 44s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60ca4089bf394801a97d862a5a05e804 : SUCCESS in 5m 33s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/76100e94851045acb9687191dec96227 : SUCCESS in 15m 17s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/7fcf0890791d45d7aeb988c6795035c5 : SUCCESS in 17m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b8f4cf776ca47a0b90cc205197bf0cb : SUCCESS in 15m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62b1cdbefe9a4d1aa3993f622b98bcd1 : SUCCESS in 17m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f525bfbab744b4b815c2dabd0f26bfd : SUCCESS in 1h 01m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/651c90f435134070b86dd111c767eac8 : SUCCESS in 1h 40m 08s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/8d8b2d5fc0af40bead172432912902b4 : SUCCESS in 24m 48s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2f02d0a99e94499ea78c1660a13aa3fb : FAILURE in 11m 21s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/695e1984569b4d15a1a8da9f676f9ee5 : FAILURE in 10m 02s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5c9407e719c64cc2a10f804fed3bfa45 : FAILURE in 11m 06s","accounts_in_message":[],"_revision_number":1},{"id":"969d5a1d15c3919bdf01b81b5c4543b3e8a97ce9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-05 16:06:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5638f84b80f92e264d68b7be3a993fa965fb0722","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 17:43:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/282fbd17fa6c42e1832338cdbde5b44c : SUCCESS in 4m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f2abd85a03184485b5b588ebd7d4bb5a : SUCCESS in 8m 08s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/de06cefac2474ff49dfb4c3b6ef559fe : SUCCESS in 6m 36s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/517819e2fa3b4cf7a96c6ffe38556627 : SUCCESS in 18m 56s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/c02e01175c204016b954785d91630857 : SUCCESS in 21m 13s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b136610088904761bc1e4b6ad2f2a214 : SUCCESS in 16m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02f1b75d7517460baf416ed9e9c5e84a : SUCCESS in 19m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aadcdcb270e54d119497063f7d12e693 : SUCCESS in 53m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/57d978da674c4eb09b23b0d043f8388a : SUCCESS in 1h 12m 56s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/67a29b7fc03d42fe9e8421796f27e54c : SUCCESS in 27m 06s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a9f0afdb1c81430582a623d752aff65b : FAILURE in 8m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dafdbcdb8c1342abbd7192d998dd506a : FAILURE in 8m 50s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eca0b803f82c45ae93663d86aab70d89 : FAILURE in 39m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ec8affb589d9ec70ca828731275e36d35e3a88df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-06 04:55:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4ec4f535ccd25d30cf3f94c8053b78653dd53ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-06 06:01:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/555687918d7b4c1aa24cf5613837203b : SUCCESS in 4m 14s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaaa8510a79b4967bf52f09e1b980a16 : SUCCESS in 4m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1fff52dbb99b45519779e09c58825d2e : SUCCESS in 5m 16s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/30ca384c1be249aa97f0cdb714af1e94 : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/f626beb9c5c94ae6a2f633e18f5fb850 : SUCCESS in 17m 39s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e831e9287c8428e97325e1f0f6b4a5d : SUCCESS in 13m 26s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38397254a2654b6da5a85bb8015a08c2 : SUCCESS in 19m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/452f0473a31c42cbb69f050d36623848 : SUCCESS in 46m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f2b3421c486416486dbc3924bff1e16 : SUCCESS in 1h 04m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/17d14a06ad094339828dd6f7fb0a606c : SUCCESS in 26m 25s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/20ef382d6b704c50a9c6e10cdf52093a : FAILURE in 15m 09s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2cbb71e943f41fe9d4d34218897d893 : FAILURE in 6m 18s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/84ae0a71bc85454abe3881c8a66f203b : FAILURE in 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"06543306f9cd9f7484143dda1dd24344dbadab0e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-09 02:06:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"68d422dbf096fa80569d556fa64e70c3690805f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 03:08:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28c8829ba7274ffeb320924b13df02d4 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8b29dc51ca744ffb8b2a960f3cbf189 : SUCCESS in 5m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d94eac177abd404eb5358fe291f53d41 : SUCCESS in 5m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/81d13f1218744f3892d5b62957901605 : SUCCESS in 18m 43s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/40eb7b10b66643f1b40e56037389a4f8 : SUCCESS in 15m 21s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6584097ad1bc49f8b5704df25b17f28e : SUCCESS in 15m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/222c7a8dca6a446986335dff6536b8e8 : SUCCESS in 18m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1bffaa995144c90a156d143fbd8392e : SUCCESS in 46m 34s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9fbf1146826f4ae29f4839f0153978b8 : SUCCESS in 1h 00m 51s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1131c136312e48c7a5efd5f3b91160f9 : SUCCESS in 29m 20s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/8446e3da164346b8b564fd8e67f64dc4 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ebbfe83d9424fcc8a40791ee5d0f15b : FAILURE in 6m 16s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c447b3bd525a422f883308733c922c02 : FAILURE in 23m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb7b1f6d4f3646a8059186c87db38a6303da99e5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-09 06:38:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b03d33b991882e62083093a86f277b0afacaa8d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 06:38:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d5218da754f84f5a774a5bbfdfb1fd5c2f5e0e58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 08:12:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dbe4cd44acd1420dbe6e4ec8345150dd : SUCCESS in 8m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cb9728e07bf84014864d857145c6f44a : SUCCESS in 6m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b60a4428fe8d4d7d9939c9e78c05a95c : SUCCESS in 7m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e07227ae50f54260a15975dfe34da52f : SUCCESS in 22m 25s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/2ca84bb2d83640cba8528a32ca7b16d6 : SUCCESS in 23m 29s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/088245f1cc9f4d92a510ab5fd1b9f180 : SUCCESS in 19m 04s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea43c00de17f4b4d9bcf6034db66e55f : SUCCESS in 55m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a28e7363469947c799b13c1757134a1f : SUCCESS in 1h 29m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7918d594e09143638dea22bf821bb74c : SUCCESS in 34m 00s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df6ab51ad9d141d783f24ca2d614b7af : FAILURE in 25m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1c41e5a6696452756985ab9bb55f0cc2b8e0415b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-09 08:12:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2bd11020e3b4068aeea934f8ce495c6b338e9a74","revisions":{"2bd11020e3b4068aeea934f8ce495c6b338e9a74":{"kind":"REWORK","_number":1,"created":"2020-04-28 14:59:19.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/34/724034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/34/724034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/724034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/724034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/34/724034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/34/724034/1"}}},"commit":{"parents":[{"commit":"c88f926c6668509e4aefb54100fa19e055c276ca","subject":"Prepare Ussuri RC1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c88f926c6668509e4aefb54100fa19e055c276ca"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:59:17.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-04-28 14:59:17.000000000","tz":0},"subject":"Update .gitreview for stable/ussuri","message":"Update .gitreview for stable/ussuri\n\nChange-Id: I3c0344b6a202607e3d3b1380c9296ea113f5f670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2bd11020e3b4068aeea934f8ce495c6b338e9a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2bd11020e3b4068aeea934f8ce495c6b338e9a74"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
