)]}'
{"id":"openstack%2Fpuppet-swift~724183","triplet_id":"openstack%2Fpuppet-swift~stable%2Ftrain~I7ce89c08c8bb82442795215a22799480f7c855b6","project":"openstack/puppet-swift","branch":"stable/train","topic":"openstacklib","hashtags":[],"change_id":"I7ce89c08c8bb82442795215a22799480f7c855b6","subject":"Add missing openstacklib to dependencies","status":"MERGED","created":"2020-04-28 22:11:36.000000000","updated":"2020-05-21 13:50:15.000000000","submitted":"2020-05-21 13:50:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724183-1590069015818-f6fe6744","meta_rev_id":"af96863b92e96c4d2a415c4cfecf47cda6f12523","_number":724183,"virtual_id_number":724183,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 13:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-21 13:39:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-21 13:02:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-21 13:39:36.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-21 13:02:15.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-21 13:39:36.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-21 13:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddda93f7aef9f333acb39b665c68d314ab061561","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-04-28 22:11:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6254a016c52ecef4f72197084830f8de526a4975","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 22:23:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/0600b3778d4a4da6aa6d3472503d5d81 : SUCCESS in 3m 40s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e76489871654137a952e10d23866999 : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"90d770ea1a71f57ffeb6a7b27afc89d604cb5d3e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-29 14:15:26.000000000","message":"Patch Set 1: Code-Review-2\n\nBecause we are just creating ussuri branch, please hold on this backport for a while.\nI\u0027m afraid that some of the existing backport to train is not included in stable/ussuri, and we should backport them to ussuri, too.","accounts_in_message":[],"_revision_number":1},{"id":"0247e7a7a56d761af06082dd3aafb8b0d4081d66","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-04-30 10:46:46.000000000","message":"Patch Set 1: Code-Review-1\n\nBecause we already have ussuri branch and it doesn\u0027t include this patch, we should backport this to ussuri first.\nPlease note that currently gate job for ussuri is mostly broken because of tripleo standalone job failng in ussuri, so please wait until the issue is fixed by tripleo side.","accounts_in_message":[],"_revision_number":1},{"id":"d25aac02f441538957a52c1cce91a0d77b8d40f7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-05 16:13:22.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nFYI, I submitted the backport of the same patch to stable/ussuri.\n\nhttps://review.opendev.org/#/c/725645/","accounts_in_message":[],"_revision_number":1},{"id":"7002ef8e41559c89826fde22d9b8b8d309ed958f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 13:01:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c6a5a95defeef6cd120a318dc0ff2278db15d707","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 13:01:36.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"eeaca3584d3a3666a0d1717f698167ea60fdd5a3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 13:02:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"78defe8f6deab65b46679f0ed91eca61bd8f4708","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-21 13:02:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b81b2cf7f33450a0df96aa09040ae6ff17b11b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:14:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/833693d455e54fca85cd20d4474c91f8 : SUCCESS in 4m 46s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6669152523f49b7939b7f3fcf0ab08d : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":2},{"id":"319d1679a113b930e4996380f05d498fb2275433","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-21 13:39:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"78f3e60cb595ad5dc2f89a7ddd4700dd64995499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:39:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"319402f1cea29ac95a10739c9212e161783b9a45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:50:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/7d8fa944ed784bff9448a35625d75f99 : SUCCESS in 2m 45s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/946ab53fa839416da9e45e54f023e355 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"874b15bc0d8d52074f48e512bf54a90f7e44e292","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:50:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3bbd1da6028f5222ea8eae61e515a6486e5af977","revisions":{"f0cfc28491371a5a06ce298f46db0b7ecbf0d75a":{"kind":"REWORK","_number":1,"created":"2020-04-28 22:11:36.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/83/724183/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/83/724183/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/1"}}},"commit":{"parents":[{"commit":"1795fc0bd04d5cb70582518bfb027c0957bc525d","subject":"Add symlink middleware support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 12:23:45.000000000","tz":540},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-04-28 22:11:20.000000000","tz":600},"subject":"Add missing openstacklib to dependencies","message":"Add missing openstacklib to dependencies\n\nChange-Id: I7ce89c08c8bb82442795215a22799480f7c855b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f0cfc28491371a5a06ce298f46db0b7ecbf0d75a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f0cfc28491371a5a06ce298f46db0b7ecbf0d75a"}]},"branch":"refs/heads/stable/train"},"3bbd1da6028f5222ea8eae61e515a6486e5af977":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-21 13:02:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/83/724183/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/83/724183/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/83/724183/2"}}},"commit":{"parents":[{"commit":"1795fc0bd04d5cb70582518bfb027c0957bc525d","subject":"Add symlink middleware support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1795fc0bd04d5cb70582518bfb027c0957bc525d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-04-25 12:23:45.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-21 13:02:08.000000000","tz":0},"subject":"Add missing openstacklib to dependencies","message":"Add missing openstacklib to dependencies\n\nChange-Id: I7ce89c08c8bb82442795215a22799480f7c855b6\n(cherry picked from commit 816f722d4e532edf9d6d1d726fd5957326eaec47)\n(cherry picked from commit 7fbf2bd2ee4f1594a4a5341b0584414247169688)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3bbd1da6028f5222ea8eae61e515a6486e5af977"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3bbd1da6028f5222ea8eae61e515a6486e5af977"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
