)]}'
{"id":"openstack%2Fpuppet-swift~724579","triplet_id":"openstack%2Fpuppet-swift~master~I3a32917f910431d38faf3607a6750fcd11d5cbea","project":"openstack/puppet-swift","branch":"master","topic":"bump-openstackdocstheme","hashtags":[],"change_id":"I3a32917f910431d38faf3607a6750fcd11d5cbea","subject":"Bump openstackdocstheme to 1.30.0","status":"MERGED","created":"2020-04-30 07:29:25.000000000","updated":"2020-04-30 08:00:18.000000000","submitted":"2020-04-30 07:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724579-1588233502737-2ccb4115","meta_rev_id":"a9c833192c3bd80194309af0993d7518746b3546","_number":724579,"virtual_id_number":724579,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 07:58:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2020-04-30 07:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2020-04-30 07:49:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 07:58:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"de2040b45dfb433373f9d1ad654d48534ea79501","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:29:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44afb8643f1bfa48002673ffa422ad30210fd869","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:33:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5384e24f31184d58875ab2ee2b8fc8f3 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9c15891ddc938281fe6ff8e4436188a38f8d0bf4","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-04-30 07:49:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e9a8b0320747321b22478823713e75ff9c243dd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:49:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"40a0103e00a3fb171647f270f4bb10615ec82710","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:58:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f4006cd124e465eb0fa5a7aa72984c7 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0bbf42da49b219cdfb19a6010a368013bad79d91","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 07:58:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c65da871210bae8777d47f3e58792816fd905438","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:00:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/948d414bdadb470c92a6199d6e914164 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d054a051f1c6969b87e4f3c479e94d3375a5a23","revisions":{"3d054a051f1c6969b87e4f3c479e94d3375a5a23":{"kind":"REWORK","_number":1,"created":"2020-04-30 07:29:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/79/724579/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/79/724579/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/724579/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/724579/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/79/724579/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/79/724579/1"}}},"commit":{"parents":[{"commit":"53fa2e738c9dd1f8f8447e6348f78997471f25bc","subject":"Change rpsec to rspec","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/53fa2e738c9dd1f8f8447e6348f78997471f25bc"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-04-30 07:28:37.000000000","tz":480},"subject":"Bump openstackdocstheme to 1.30.0","message":"Bump openstackdocstheme to 1.30.0\n\n...to pick up many improvements, including the return of table borders.\n\nChange-Id: I3a32917f910431d38faf3607a6750fcd11d5cbea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d054a051f1c6969b87e4f3c479e94d3375a5a23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d054a051f1c6969b87e4f3c479e94d3375a5a23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
