)]}'
{"id":"openstack%2Fpuppet-swift~725119","triplet_id":"openstack%2Fpuppet-swift~master~I7a4010f833da12f7adbdae8cfc343d0fb39b1552","project":"openstack/puppet-swift","branch":"master","hashtags":[],"change_id":"I7a4010f833da12f7adbdae8cfc343d0fb39b1552","subject":"Cleanup deprecated parameters","status":"MERGED","created":"2020-05-03 15:29:10.000000000","updated":"2020-05-04 02:07:11.000000000","submitted":"2020-05-04 02:05:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"725119-1588557959779-56a7dbf5","meta_rev_id":"6b5b3ae9d0c2d9c18b1c0c6e0ec9f224dbe37fea","_number":725119,"virtual_id_number":725119,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-04 02:05:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-04 00:36:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-03 19:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-04 00:36:07.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-03 19:59:46.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-04 00:36:07.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-04 02:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4405abc8b1436c919562bc9e99f8e44acdee5af1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 15:29:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd488f3bc6c2d7a9900fbe35349b81c095be7694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 16:59:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1141c78c89bc4c8c98ce65f91121fc1a : SUCCESS in 4m 52s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fefb02ca0b6b4e38a6d2a3d810d2ff86 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/20e6b4e0ad35438aa15ee25cb4d8884d : SUCCESS in 4m 53s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d13b43d3ec364496a9da75376004bb69 : SUCCESS in 21m 33s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/622232e51ff34e81b6574fa02664b2e8 : SUCCESS in 16m 35s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8a4a6f4427b4071b2ea862177fa6b36 : SUCCESS in 13m 51s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/999dd9b91cd14705b292d0081debcf3b : SUCCESS in 16m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6e2670c434e2436890d68e07b6e1e4b3 : SUCCESS in 58m 37s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5183dea94b18461f9292b422d08a2612 : FAILURE in 1h 01m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9214050e56994c0c9a29323a391a4a5d : FAILURE in 26m 17s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b7729d42628c4e4f9d333a2dd544c87e : FAILURE in 10m 55s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/506a3248c6a24d1eb82d3cab1e06b23d : FAILURE in 7m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ec66c41a18748f289acf1e337f049bc : FAILURE in 3m 27s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18e5c931624246128618f8d9c3f9a04c : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"35068bee29dca63c30deb10139a2fe4f70414daf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 17:36:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"57c7f9779ab171cb54620be964d21cd4b6483526","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-03 18:02:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f74798f1d7b9a1fdee8bd9e49ad9f611ef90a654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 19:50:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df085532e6d14b4cad5675431e3ef002 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2f668901f6cf4b7a96b68b97a3347fa7 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aafba4b0cd3472fa70ddd0b21f4de15 : SUCCESS in 5m 41s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2b51051273464bcea00be3727fbe8807 : SUCCESS in 18m 18s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/511e28a7c09f4568a4abab35c53be374 : SUCCESS in 16m 59s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2abc5c1d49024810ba0470aa63225267 : SUCCESS in 18m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2d35ae597217425cab723ea0179dbef4 : SUCCESS in 19m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da74c7225c0484f82433e04ec3355c4 : SUCCESS in 1h 01m 26s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/615ec1cbb3b04433a21c88611bbafbea : SUCCESS in 1h 13m 01s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2816863f04b94076b430eed6193f8b29 : SUCCESS in 30m 24s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/f1742823250743ddae3b6739dee8e316 : FAILURE in 11m 49s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10e4c18678fd4a37a18b475a83b9a61b : FAILURE in 9m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21378794e42042929f79fb3dc7295dd1 : SUCCESS in 3m 58s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18f6f4f7ef564a0db7814d39ac49a900 : SUCCESS in 1h 39m 16s","accounts_in_message":[],"_revision_number":3},{"id":"d1e6d10a06894df351db9bd29ae0abfc4e80dcc5","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-03 19:59:46.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"012a6ef28e2fe75381b70433ba519d51607c5bda","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-04 00:36:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d47d5c87f7b62f5ae2b51749b517d7eafb0fb758","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 00:36:19.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd47d73063b6943d57b496b7ebdc14f6dad7d3cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:05:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4135d21d05ed48adaff94a1cf50dfefb : SUCCESS in 4m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba5c258886d5426497401130040144c9 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72d369f4b90d42ceb0b7187b55fae161 : SUCCESS in 3m 52s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/79b0daca64df4c26981fca18231b0e9a : SUCCESS in 21m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b06014031c45496cbd3feaf1bd7ed4ad : SUCCESS in 19m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ee20d4039b84b1aba4bfc996c0516b3 : SUCCESS in 17m 47s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8609b42537c946ff9d99fd25cb588149 : SUCCESS in 49m 25s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/af695dc9815247dab2fb9a14250d1a2e : SUCCESS in 1h 01m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a3d3847c8eb444cd88aa662e59177dbb : SUCCESS in 32m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d95dc53420c649cfac341815a9b6ea58 : SUCCESS in 3m 24s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cd70370655c54349980917062dc2097e : SUCCESS in 1h 27m 42s","accounts_in_message":[],"_revision_number":3},{"id":"ef210e62db331ee0f2fd3a4ca3a327bc637ae9c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:05:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"6b5b3ae9d0c2d9c18b1c0c6e0ec9f224dbe37fea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-04 02:07:11.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e5d16e093194033bb40d35db02414f9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"56b76020af3761481cc3b0bf5bfb2056970d2992","revisions":{"731ba98ddde423e8d33c1f4a1e35dd97aac4052b":{"kind":"REWORK","_number":1,"created":"2020-05-03 15:29:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/725119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/19/725119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/1"}}},"commit":{"parents":[{"commit":"a74a03cc4746524ba3161c2a18316ea1a96fc7b6","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a74a03cc4746524ba3161c2a18316ea1a96fc7b6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:28:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:28:52.000000000","tz":540},"subject":"Cleanup deprecated parameters","message":"Cleanup deprecated parameters\n\nChange-Id: I7a4010f833da12f7adbdae8cfc343d0fb39b1552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/731ba98ddde423e8d33c1f4a1e35dd97aac4052b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/731ba98ddde423e8d33c1f4a1e35dd97aac4052b"}]},"branch":"refs/heads/master"},"24da53c3e196fe1f832f1c572b2ebff9cb86205a":{"kind":"REWORK","_number":2,"created":"2020-05-03 17:36:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/725119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/19/725119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/2"}}},"commit":{"parents":[{"commit":"a74a03cc4746524ba3161c2a18316ea1a96fc7b6","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a74a03cc4746524ba3161c2a18316ea1a96fc7b6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:28:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 17:36:05.000000000","tz":540},"subject":"Cleanup deprecated parameters","message":"Cleanup deprecated parameters\n\nChange-Id: I7a4010f833da12f7adbdae8cfc343d0fb39b1552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/24da53c3e196fe1f832f1c572b2ebff9cb86205a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/24da53c3e196fe1f832f1c572b2ebff9cb86205a"}]},"branch":"refs/heads/master"},"56b76020af3761481cc3b0bf5bfb2056970d2992":{"kind":"REWORK","_number":3,"created":"2020-05-03 18:02:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/19/725119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/19/725119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/19/725119/3"}}},"commit":{"parents":[{"commit":"a74a03cc4746524ba3161c2a18316ea1a96fc7b6","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a74a03cc4746524ba3161c2a18316ea1a96fc7b6"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 15:28:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-03 18:01:52.000000000","tz":540},"subject":"Cleanup deprecated parameters","message":"Cleanup deprecated parameters\n\nChange-Id: I7a4010f833da12f7adbdae8cfc343d0fb39b1552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/56b76020af3761481cc3b0bf5bfb2056970d2992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/56b76020af3761481cc3b0bf5bfb2056970d2992"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
