)]}'
{"id":"openstack%2Fpuppet-swift~727190","triplet_id":"openstack%2Fpuppet-swift~master~I6e7721d04ae2bf2e2a2ea3f02ebfcbded58692e2","project":"openstack/puppet-swift","branch":"master","topic":"password","hashtags":[],"change_id":"I6e7721d04ae2bf2e2a2ea3f02ebfcbded58692e2","subject":"Deprecate the default values for password parameters","status":"MERGED","created":"2020-05-12 13:29:01.000000000","updated":"2020-05-13 18:10:16.000000000","submitted":"2020-05-13 18:08:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727190-1589393312937-8773121f","meta_rev_id":"e76cdb883241150a9110b0c079e8bb7e1e832b5e","_number":727190,"virtual_id_number":727190,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 18:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-05-13 16:13:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-13 06:50:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-05-13 16:13:46.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 06:50:08.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-13 16:13:46.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-13 18:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ef577c85f130f2aaac38ac14cb73c234be036c5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 13:29:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7572dc074896c6f93654891c27e0e9408c1f09","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-12 13:32:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e8c30df8f46b5ee850cc4ff9bd69e17de0dce4e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 13:35:22.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e53a24b4ec735049efbafc1213d8d349c8e0ccd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 13:44:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a51f79a2dbbd7924de56cf62c6776c7618d0949b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-12 13:53:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eda0cbba9f0539aaa2109467c64a0e4b82f41ab4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 15:16:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e2058173e85a41a2bc873f4e66f49663 : SUCCESS in 5m 44s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef91f6516bcd4872b33f1851bb43462f : SUCCESS in 5m 38s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d948f64ecedc4f28842ac886b4052f3b : SUCCESS in 6m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6942a03385394fb5a82f217acedcc391 : FAILURE in 19m 31s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8f0e0d1032c647fc84026b2e085174bf : FAILURE in 18m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/914b837f4c894cc89022690e95abaa5c : FAILURE in 17m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1166683ec514b24b6cd24e9c502718d : FAILURE in 19m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd3599517eec4d69be6f70dde0850e04 : SUCCESS in 48m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/d01eda214a1f4447be56156462c811ef : SUCCESS in 1h 06m 52s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ab202270e92545fb91a60b71f8d471d4 : SUCCESS in 27m 24s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6b0b83776e034f1096c2f7202f0c1bb3 : FAILURE in 9m 53s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd925fcabe374616a001e7f357906d60 : FAILURE in 7m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c38db24bcd7c4d51bbf69abd5d47a578 : SUCCESS in 5m 23s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b1b0d4755f8e46e7a6eaa736d91b1d24 : SUCCESS in 1h 29m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d8b4bb60e3fec931d0599e83a0340a6c3a0d99d5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-12 23:53:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8c2ce731622c72d758d895fe51cd4df64f85d594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 01:24:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd5a0aafe2054fb8a5df9db97dc44ac2 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef1df91b8ce1424d8b3ec5cc6e1e6a1b : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9b8114535cb648599bc4239db4e5df18 : SUCCESS in 4m 51s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/1a9c1a6ab8734a03b3e824adfc7683b9 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3b4dc07fcc844a73912a79cd4c2f90d2 : SUCCESS in 17m 23s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52e9b9e7d42a400dba5e64a99284eeec : SUCCESS in 14m 36s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18f422f627874bef9829af36a4b2be22 : SUCCESS in 20m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73348970823b407e9eb59b470a5cd9b6 : SUCCESS in 55m 20s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/12b22c72d7e14126a9dcee0576beb458 : SUCCESS in 1h 02m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/5d335c9b99c54f4dadb59683b25748b2 : SUCCESS in 28m 09s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2e284d635d694a6da26f7c3f9ed99a67 : FAILURE in 8m 12s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0a317420b6b43758a6650a94b7a34b4 : FAILURE in 7m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49e42d6cbf064a6899813b3dc791220d : SUCCESS in 5m 03s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eda248f3819e4135b0771b35f4f78f27 : SUCCESS in 1h 24m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2ecdc22ff843107654e01655be1c106d11ae936f","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-13 06:50:08.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3855a84b52b5ef5af890beac5185e99bcd4759df","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-13 16:13:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aa083a88960aa3992354fc40bbb4363bd17a2a39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 16:17:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de103df48ea2c35566c924a22473035964712476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:08:32.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bb9e7875273048de8f1db8c6957d29d1 : SUCCESS in 6m 20s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6669a6eedc94b66980e8b8ea52af985 : SUCCESS in 4m 55s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c5aa4c90c8504ddc97e2a09e0739493c : SUCCESS in 5m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ef7002117c994ae887ad4867edbdc019 : SUCCESS in 16m 25s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/1e8779dab9044ac489bf712327d9b1b4 : SUCCESS in 18m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0c555d385f14385ae12fb96081f05a1 : SUCCESS in 18m 39s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/412e01665e7d4ff2824721ee31fe1efe : SUCCESS in 46m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ce20cacbbfb94dd3b1db8e1f485a7d14 : SUCCESS in 1h 17m 12s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bac9dfb87e3c421eacfa1d336dca0b65 : SUCCESS in 31m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984d4c2575834f629e0dbca704278bf3 : SUCCESS in 4m 02s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2ee7e5cac8214b4b8152b14e97e68aa3 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":3},{"id":"ae0f4054f4f41572cbaa35fdc10329e104390945","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:08:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"b26f055bc6ce984a30551ac66f739cba97251c64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:10:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/732441359470474aa14ca113522ef388 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2df992faf3d2a6c845b6bbe302fb956b77307c50","revisions":{"7935209ef9edb7966cd70250030d0ffa17cb4e51":{"kind":"REWORK","_number":1,"created":"2020-05-12 13:29:01.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/727190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/727190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/1"}}},"commit":{"parents":[{"commit":"de443f28fc95b07d78c1761c3deefce2da916eaa","subject":"Passwords should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de443f28fc95b07d78c1761c3deefce2da916eaa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 13:27:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 13:27:12.000000000","tz":540},"subject":"Deprecate the default values for password parameters","message":"Deprecate the default values for password parameters\n\nCurrently puppet-swift provides default values for some password\nparameters, but this is not ideal from security perspective and we\nshould expect operators to set their own password explicitly.\n\nThis patch deprecates the usage of these default values and adds\nwarning message which appears for missing password defined, so that\nwe can remove current default values in next cycle.\n\nChange-Id: I6e7721d04ae2bf2e2a2ea3f02ebfcbded58692e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7935209ef9edb7966cd70250030d0ffa17cb4e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7935209ef9edb7966cd70250030d0ffa17cb4e51"}]},"branch":"refs/heads/master"},"859666d9195c237e7a8499cb51a725f23baaae3c":{"kind":"REWORK","_number":2,"created":"2020-05-12 13:44:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/727190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/727190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/2"}}},"commit":{"parents":[{"commit":"de443f28fc95b07d78c1761c3deefce2da916eaa","subject":"Passwords should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de443f28fc95b07d78c1761c3deefce2da916eaa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 13:27:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 13:34:49.000000000","tz":540},"subject":"Deprecate the default values for password parameters","message":"Deprecate the default values for password parameters\n\nCurrently puppet-swift provides default values for some password\nparameters, but this is not ideal from security perspective and we\nshould expect operators to set their own password explicitly.\n\nThis patch deprecates the usage of these default values and adds\nwarning message which appears for missing password defined, so that\nwe can remove current default values in next cycle.\n\nChange-Id: I6e7721d04ae2bf2e2a2ea3f02ebfcbded58692e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/859666d9195c237e7a8499cb51a725f23baaae3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/859666d9195c237e7a8499cb51a725f23baaae3c"}]},"branch":"refs/heads/master"},"2df992faf3d2a6c845b6bbe302fb956b77307c50":{"kind":"REWORK","_number":3,"created":"2020-05-12 23:53:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/727190/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/727190/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/727190/3"}}},"commit":{"parents":[{"commit":"de443f28fc95b07d78c1761c3deefce2da916eaa","subject":"Passwords should be secret","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/de443f28fc95b07d78c1761c3deefce2da916eaa"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 13:27:12.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-05-12 23:53:14.000000000","tz":540},"subject":"Deprecate the default values for password parameters","message":"Deprecate the default values for password parameters\n\nCurrently puppet-swift provides default values for some password\nparameters, but this is not ideal from security perspective and we\nshould expect operators to set their own password explicitly.\n\nThis patch deprecates the usage of these default values and adds\nwarning message which appears for missing password defined, so that\nwe can remove current default values in next cycle.\n\nChange-Id: I6e7721d04ae2bf2e2a2ea3f02ebfcbded58692e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2df992faf3d2a6c845b6bbe302fb956b77307c50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2df992faf3d2a6c845b6bbe302fb956b77307c50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
