)]}'
{"id":"openstack%2Fpuppet-swift~727918","triplet_id":"openstack%2Fpuppet-swift~master~I3992008c88e8a268bd507e356b6166aed34a60f6","project":"openstack/puppet-swift","branch":"master","topic":"deps","hashtags":[],"change_id":"I3992008c88e8a268bd507e356b6166aed34a60f6","subject":"Bump compatibility versions","status":"MERGED","created":"2020-05-14 04:02:27.000000000","updated":"2020-05-18 23:06:22.000000000","submitted":"2020-05-18 23:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727918-1589843182557-9592ee8f","meta_rev_id":"48840b73d19b2f663345aa9d8cb65d36a4dfc322","_number":727918,"virtual_id_number":727918,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-14 10:20:24.000000000","_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 23:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":2,"date":"2020-05-18 22:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-14 17:42:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":1,"date":"2020-05-18 22:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-14 17:42:09.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-18 22:58:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-18 23:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e42e3949c27114bc0707b987bad9119977823a42","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-14 04:02:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2333e5d10851f7708c969349fe149bd946dd05c","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-14 04:02:54.000000000","message":"Topic changed from ups/master to deps","accounts_in_message":[],"_revision_number":1},{"id":"f8f8d26b958fa29da6655959a885dc785f463dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 04:11:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/cbfa77fae4ef4bc3969631dbf74cf836 : SUCCESS in 2m 03s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9460fd03453a469995dad8efd565e4db : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d8895467641359b05a85b7581f2eef08027244ed","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 08:45:04.000000000","message":"Patch Set 1: Code-Review-1\n\nWe have not tested with that concat version, we are pinned to 2.2.1 [1]. If we should bump this we first need to ensure our testing (and everything works) with later concat versions.\n\n[1] https://github.com/openstack/puppet-openstack-integration/blob/master/external_modules.txt#L8","accounts_in_message":[],"_revision_number":1},{"id":"d99bf38c73c1158e590c14288ff83ed0939ff64b","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-14 10:20:24.000000000","message":"Patch Set 1:\n\n\u003e We have not tested with that concat version, we are pinned to 2.2.1\n \u003e [1]. If we should bump this we first need to ensure our testing\n \u003e (and everything works) with later concat versions.\n \u003e \n \u003e [1] https://github.com/openstack/puppet-openstack-integration/blob/master/external_modules.txt#L8\n\nI understand your test does not include other versions. \n\n \u003e We have not tested with that concat version, we are pinned to 2.2.1\n \u003e [1]. If we should bump this we first need to ensure our testing\n \u003e (and everything works) with later concat versions.\n \u003e \n \u003e [1] https://github.com/openstack/puppet-openstack-integration/blob/master/external_modules.txt#L8\n\nThanks, so I need to submit a patch for the version bump in puppet-openstack-integration, right?  The reason I submit this is because I found some versions are quite old and the dependencies are actually broken. One example is swift requires stdlib: \"\u003e\u003d5.0.0 \u003c7.0.0\" but latest rsync which is required by swift requires stdlib:  \"\u003e\u003d4.2.0 \u003c5.0.0\"","accounts_in_message":[],"_revision_number":1},{"id":"b98200beb3f0881219510bab9cc1c36aa5463e3e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 17:42:09.000000000","message":"Patch Set 1: Code-Review+2\n\nThis is fine according to testing in https://review.opendev.org/#/c/728068/","accounts_in_message":[],"_revision_number":1},{"id":"95059fa53a61653ae3b8e63d9f3451cf23c772bb","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-14 17:43:00.000000000","message":"Patch Set 1:\n\nYea, I\u0027ve wanted to fix up a job that installs with modules from PuppetForge just to validate the dependency setup in metadata and then only use the actual module that was changed in jobs but haven\u0027t had time to look at that.","accounts_in_message":[],"_revision_number":1},{"id":"289d341d5bced35d35120f4ed845178f11fc21a1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-18 22:58:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40d0c8f6894aadc1c058c48110a2c4c456c5654c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 22:58:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fd8759568939f9adc65e65b299fd99c9abf43af1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:06:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"48840b73d19b2f663345aa9d8cb65d36a4dfc322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 23:06:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-puppet-tarball https://zuul.opendev.org/t/openstack/build/50db1d390404489bb9931cdb565c8e53 : SUCCESS in 2m 01s\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17d4459a373a4aa4815a0e6c39441e72 : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"930c178873f57b7ce2a16fd58d724ffea56b6bf7","revisions":{"930c178873f57b7ce2a16fd58d724ffea56b6bf7":{"kind":"REWORK","_number":1,"created":"2020-05-14 04:02:27.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/18/727918/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/18/727918/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/727918/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/727918/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/18/727918/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/18/727918/1"}}},"commit":{"parents":[{"commit":"b753df974b2892352da415438680b06b2dc07513","subject":"Merge \"Accept integer for port numbers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b753df974b2892352da415438680b06b2dc07513"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-14 04:00:47.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-14 04:01:02.000000000","tz":600},"subject":"Bump compatibility versions","message":"Bump compatibility versions\n\nChange-Id: I3992008c88e8a268bd507e356b6166aed34a60f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/930c178873f57b7ce2a16fd58d724ffea56b6bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/930c178873f57b7ce2a16fd58d724ffea56b6bf7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
