)]}'
{"id":"openstack%2Fpuppet-swift~728213","triplet_id":"openstack%2Fpuppet-swift~master~Ib5262c26244363a1c47bdd562d560735ddc82a36","project":"openstack/puppet-swift","branch":"master","topic":"fix-mount-options","hashtags":[],"change_id":"Ib5262c26244363a1c47bdd562d560735ddc82a36","subject":"Fix disk mount options","status":"MERGED","created":"2020-05-14 18:16:18.000000000","updated":"2020-05-21 18:40:43.000000000","submitted":"2020-05-21 18:38:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728213-1590086302893-2e69b6b6","meta_rev_id":"3670c396650c44222ddd081a39dd2311a2cb4ad7","_number":728213,"virtual_id_number":728213,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-21 18:38:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-19 14:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-21 16:49:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2020-05-21 16:49:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-15 07:41:54.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-19 14:51:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-21 16:49:40.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-05-21 18:38:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"88ad2c1b29b514356961cb2fef8228c942ddb316","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-14 18:16:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f9d54c28a3b24248765a6724b9dcdc0d56cd93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 20:08:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6131be6d10d84614bb1dfabcd16bd0b0 : SUCCESS in 7m 03s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e05f1059ce04a17818f11d807130632 : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/352f9e79bee6483882b7a9e1590831ff : SUCCESS in 7m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4b1204d45ffb41a0ba56f18b8232b9b9 : FAILURE in 23m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/597668e7135a40d0b7d7a78b4f9541dc : FAILURE in 18m 49s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1c32042b54c4e63b156d99864dd2056 : FAILURE in 17m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/36d6520fdd1e4ce891b602e94861071c : FAILURE in 22m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47da4a1f540f4889916beaeab8ede9ca : SUCCESS in 52m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/89b4cd77e35c41fa815af17c6a85c0cf : SUCCESS in 1h 03m 27s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/fa1441739631480d94643c6e44491fe1 : SUCCESS in 28m 59s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e4bee0ad079d4531a79983d886267340 : FAILURE in 12m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7d75dd6c1754a71a586a2ae7ec75807 : FAILURE in 7m 13s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a56ab25bb550407dbeea3093c9a53ba7 : SUCCESS in 1h 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"096983bff4f5ccd1508dbc6e73b1cb7a9465c910","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-14 20:57:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1a4849fedf699570430823ab832e155dc22c041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 22:20:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/714d3a98657b43c9bbf06906bacd7494 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e43c0e9bbfbf42a58fb6045aa8fa7263 : SUCCESS in 4m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0e5b72b607b4473994f7195c90109c9 : SUCCESS in 5m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/2eb520a1e07747928d5bee3bb8466003 : SUCCESS in 18m 26s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0d7955b928734aa6b6b86438c3a24253 : SUCCESS in 16m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12dc8b03eab445a4981ccf16a7e092d2 : SUCCESS in 15m 08s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2aa1abe58b154942b73f568ab953b18e : SUCCESS in 16m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41f9b7a10ebd4c068a2098b0d09a4107 : SUCCESS in 50m 03s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b71ceceb660d49ce8ccf5a3fdb17cca6 : FAILURE in 1h 08m 26s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/23961f2e647349369b213a6d59174dae : SUCCESS in 25m 22s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/a32014ba3a1b4ac98a6aed930f9d590d : FAILURE in 9m 40s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/513011ecbcbd4d4ea323bc735c2303a0 : FAILURE in 7m 47s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/04d395cfefc2440fa0ea85c0f762feba : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":2},{"id":"f0dd9b70d69becb9260e55544d9c7c1bba45a54c","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-14 22:39:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"46621c1914c928ad32b392c1b7d56ce97fa6609a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 00:15:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6bfcce677f114b4e8a706aa842f1b1bb : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b47373ca74e54efeb10ec5747f61aa37 : SUCCESS in 5m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efacda171fe2490d9ec38d907990f2c2 : SUCCESS in 6m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d4afe71289de4630bdd1009da40ef77d : SUCCESS in 17m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/7d557b742f644b3184af160682391054 : SUCCESS in 19m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da62778c91b4244aa6ff7b286524d36 : SUCCESS in 18m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bbf50d602334f06817ae5600dc8dffd : SUCCESS in 18m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e157c19ae0914dfeac1ecb92c498de7e : SUCCESS in 55m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/9bc75e3f4a6c482d82d9e13069c48386 : SUCCESS in 56m 43s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7a4b54786a2849e099091ac43222f0cc : SUCCESS in 28m 47s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/6cfc9f4633454f18ab822b366d0f339c : FAILURE in 10m 10s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f590d592fbe44f7ae6ec58a227b56a1 : FAILURE in 10m 28s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f35411c9b47a4043ac0fc2daf97527fd : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":2},{"id":"faa396eda589dc6e2b8f3e5154ba9770e7809d39","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-15 07:41:54.000000000","message":"Patch Set 2: Code-Review-1\n\nmissing a release note","accounts_in_message":[],"_revision_number":2},{"id":"7ea61f4519353bf1abd70b60383508a72e4d68e8","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-15 21:00:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"10c9d5270289fd8f034a07c023421495ce736ddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 22:30:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c495156e7d884c409f54f79e2f57d1d9 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b25e5fd6c76f4507a0d0074696d0625c : SUCCESS in 9m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62703cef98524044818e20baa25f649d : SUCCESS in 5m 19s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/11603d2973bc48ef83850e97d0de3ed6 : FAILURE in 16m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f721ede2b62e41888ab12c8db2d49818 : FAILURE in 19m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9ca755dfc304c3fba85ba72180cd59e : FAILURE in 16m 29s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9924e533300487096dd49df23566793 : FAILURE in 17m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d433acac1e8e44adb147a11561af5e18 : SUCCESS in 50m 09s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3ade870da7cf491c96d5b138f23028bb : SUCCESS in 1h 05m 11s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f2277fd23fe846c98294242c6512f2a4 : SUCCESS in 29m 23s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fc49b7e1451b472c982d36d7b4502c2d : FAILURE in 8m 22s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7bdfbb422beb4c57b3a637c99fe37cf5 : FAILURE in 7m 13s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af1367f91a314043b9576986f53d3083 : SUCCESS in 3m 40s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e261350bb1b44a1a9d695a40a0e12b36 : SUCCESS in 1h 26m 51s","accounts_in_message":[],"_revision_number":3},{"id":"0095b0f83662f4865abe00388ebf547b6886aea3","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-16 10:22:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5feba52baed61a5cfce8c39080f5a9b35b672502","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-16 11:48:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e66f2e724c7d4c6891a757d09069623e : SUCCESS in 4m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaa537e332d84ca4875f3dfb8fa04036 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f8b755ca28c48ea96c1e34ee2d3c250 : SUCCESS in 4m 18s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/584f4fb695da411584c4852b2b211510 : SUCCESS in 18m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/ed162a6671d14593b0aba1b1e4880500 : SUCCESS in 16m 04s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a096467d30594deeb7df459fcc4dd74e : SUCCESS in 15m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3f02a7515884829b84ae6a20115d8b6 : SUCCESS in 15m 45s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbf0c4ef034d41fd8b989a41fbff0470 : FAILURE in 39m 40s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b1a352dde89746a6a5faf5f6169237fa : SUCCESS in 1h 09m 36s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c85101e0bdc54e849bd565f7ec9b7eeb : SUCCESS in 28m 03s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c25a6d427db74837947ace6499845b37 : FAILURE in 9m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4222e5d7a070435aaabf0c5b5a4982a9 : FAILURE in 6m 05s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5967deca99744343a2f21f2acaecfd3c : SUCCESS in 3m 12s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/31b01466545c4de5bea46c0cfc3e6199 : SUCCESS in 1h 22m 33s","accounts_in_message":[],"_revision_number":4},{"id":"d9c08c133fa0eb4552bd02a8bccce0a59acd11d0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 13:59:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"43ceadb917f7cc6c26ffd46bcbf8930c0a8d4fb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 14:10:36.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fde1fe3ec4ec2ea02b4217aa3af785b2d4d0d466","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-19 14:37:20.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"226c5753a34fe9ad67b6fc49c2a903a537b6da68","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2020-05-19 14:38:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a0ff9ba5d4bd449393d0ff51cd05b414dafcdc62","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 14:50:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d7dd4a34618cbe97379b31438a669a611a215853","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-19 14:51:25.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"d0340dd9c5006982bd8d6e53a24b823a64d530d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 16:01:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72bd072576624269ae1b5435fca1fc20 : SUCCESS in 5m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2cd5d7f7ee6443d903430c19471860e : SUCCESS in 5m 22s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/239a306ff04049478ec7ad0b0213a01f : SUCCESS in 4m 27s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ad32cfa3786c4a7d9a74d40aa52f407b : SUCCESS in 20m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4c7b2f946dbe425da90df5dabd03f28f : SUCCESS in 18m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42670b620dfd4216a0d9e978ae848011 : SUCCESS in 17m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dd65baceeffc47dfa83a4f9b3e72e5a5 : SUCCESS in 16m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/244bb214f5de4e778eb419aba4cf331f : FAILURE in 35m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8498250b77a741058ccde89df10f1691 : SUCCESS in 1h 00m 39s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b615d469e5ec49148c7bd16c187c645c : SUCCESS in 27m 50s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dd24b6e2a0c24035a9ca0c106e999b19 : FAILURE in 9m 02s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d0638c35e9e4e959866c727fb342f00 : FAILURE in 9m 24s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17ec7645f5d64d18b9d4f7fe4cea3c72 : SUCCESS in 3m 35s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8b5b86f0998a44ff9f94ec303d59f71d : SUCCESS in 1h 20m 12s","accounts_in_message":[],"_revision_number":5},{"id":"d1305553d6ec8f1de428801e0a94e513c2c91c38","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-05-21 16:49:40.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5bf247a473ba5f6ed0cc44e6388f396382f86372","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 16:51:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"91dfb04c479d731aacee0969ca68fef40f12f739","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:38:22.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/429c29a0f6ff429c9aa6bf870e4f5375 : SUCCESS in 7m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/561aaaf7541e472c8b4bb567bdafe85a : SUCCESS in 5m 43s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/436d53abbedd4217bcecc5ea4fb2e65b : SUCCESS in 6m 48s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/c2068fd9e0f84fb69a770ef62f83fc9e : SUCCESS in 20m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/fcedcd99d9dc4f45827cb856c2f7155d : SUCCESS in 20m 15s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05f028b5ddf64872bb43277bbc1550d9 : SUCCESS in 18m 15s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43a685cebe314e5aba92d506131b99ba : FAILURE in 36m 30s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f91e3a8806a1418aad08bbf375efc4dc : SUCCESS in 1h 04m 34s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/03d3a50105a7425098feffc8f42d4983 : SUCCESS in 33m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2dfba4383c14ed692685380e5eb344d : SUCCESS in 4m 38s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a2ab2d7abeb14aaea1029a55eaae4fa7 : SUCCESS in 1h 40m 24s","accounts_in_message":[],"_revision_number":5},{"id":"c227df86356c0868c919d411055d6461dc4bf6ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:38:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"78f62e9612e63bddf7440a4ff18c8f5cd9c41d5e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 18:40:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/243d2fb2a6614d4c8c8271edd51a5bb2 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3c3a1212b802e9463b22b5941ccac5001056fde4","revisions":{"c56d04abe544011f56375cf247317953fa6405fd":{"kind":"REWORK","_number":1,"created":"2020-05-14 18:16:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/728213/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/13/728213/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/1"}}},"commit":{"parents":[{"commit":"ea1421b4ef4768a0d1662ba6c26311aa3983905c","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea1421b4ef4768a0d1662ba6c26311aa3983905c"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:14:37.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:15:52.000000000","tz":120},"subject":"Fix disk mount options","message":"Fix disk mount options\n\nThe nobarrier option is invalid in recent kernels, and having it\nmakes the mount fail.\n\nAlso, adding \"nofail\" to avoid failure when booting a server if\nthere is a broken drive, which is quite common on a large cluster.\n\nChange-Id: Ib5262c26244363a1c47bdd562d560735ddc82a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c56d04abe544011f56375cf247317953fa6405fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c56d04abe544011f56375cf247317953fa6405fd"}]},"branch":"refs/heads/master"},"c15def2592e88d071059c6910ff8d3e2bc30d388":{"kind":"REWORK","_number":2,"created":"2020-05-14 20:57:10.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/728213/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/13/728213/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/2"}}},"commit":{"parents":[{"commit":"ea1421b4ef4768a0d1662ba6c26311aa3983905c","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea1421b4ef4768a0d1662ba6c26311aa3983905c"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:14:37.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 20:56:58.000000000","tz":120},"subject":"Fix disk mount options","message":"Fix disk mount options\n\nThe nobarrier option is invalid in recent kernels, and having it\nmakes the mount fail.\n\nAlso, adding \"nofail\" to avoid failure when booting a server if\nthere is a broken drive, which is quite common on a large cluster.\n\nChange-Id: Ib5262c26244363a1c47bdd562d560735ddc82a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c15def2592e88d071059c6910ff8d3e2bc30d388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c15def2592e88d071059c6910ff8d3e2bc30d388"}]},"branch":"refs/heads/master"},"97bb8e33e0ffdd7eb95c331b1b372dbccc43711a":{"kind":"REWORK","_number":3,"created":"2020-05-15 21:00:24.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/728213/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/13/728213/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/3"}}},"commit":{"parents":[{"commit":"ea1421b4ef4768a0d1662ba6c26311aa3983905c","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea1421b4ef4768a0d1662ba6c26311aa3983905c"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:14:37.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-15 21:00:09.000000000","tz":120},"subject":"Fix disk mount options","message":"Fix disk mount options\n\nThe nobarrier option is invalid in recent kernels, and having it\nmakes the mount fail.\n\nAlso, adding \"nofail\" to avoid failure when booting a server if\nthere is a broken drive, which is quite common on a large cluster.\n\nChange-Id: Ib5262c26244363a1c47bdd562d560735ddc82a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/97bb8e33e0ffdd7eb95c331b1b372dbccc43711a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/97bb8e33e0ffdd7eb95c331b1b372dbccc43711a"}]},"branch":"refs/heads/master"},"fa5e35582bcd8013941fb6218f44f13ea5b5174f":{"kind":"REWORK","_number":4,"created":"2020-05-16 10:22:05.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/728213/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/13/728213/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/4"}}},"commit":{"parents":[{"commit":"ea1421b4ef4768a0d1662ba6c26311aa3983905c","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea1421b4ef4768a0d1662ba6c26311aa3983905c"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:14:37.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-16 10:21:51.000000000","tz":120},"subject":"Fix disk mount options","message":"Fix disk mount options\n\nThe nobarrier option is invalid in recent kernels, and having it\nmakes the mount fail.\n\nAlso, adding \"nofail\" to avoid failure when booting a server if\nthere is a broken drive, which is common on a large cluster.\n\nChange-Id: Ib5262c26244363a1c47bdd562d560735ddc82a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fa5e35582bcd8013941fb6218f44f13ea5b5174f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fa5e35582bcd8013941fb6218f44f13ea5b5174f"}]},"branch":"refs/heads/master"},"3c3a1212b802e9463b22b5941ccac5001056fde4":{"kind":"REWORK","_number":5,"created":"2020-05-19 14:38:18.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/13/728213/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/13/728213/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/13/728213/5"}}},"commit":{"parents":[{"commit":"ea1421b4ef4768a0d1662ba6c26311aa3983905c","subject":"Merge \"Use anchor to require necessary packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea1421b4ef4768a0d1662ba6c26311aa3983905c"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-14 18:14:37.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2020-05-19 14:37:27.000000000","tz":120},"subject":"Fix disk mount options","message":"Fix disk mount options\n\nThe nobarrier option is invalid in recent kernels (\u003e\u003d 4.19), and having\nit makes the mount fail. It has no effect on kernel \u003e\u003d 4.10.\n\nAlso, adding \"nofail\" to avoid failure when booting a server if\nthere is a broken drive, which is common on a large cluster.\n\nChange-Id: Ib5262c26244363a1c47bdd562d560735ddc82a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3c3a1212b802e9463b22b5941ccac5001056fde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3c3a1212b802e9463b22b5941ccac5001056fde4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
