)]}'
{"id":"openstack%2Fpuppet-swift~730688","triplet_id":"openstack%2Fpuppet-swift~master~I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e","project":"openstack/puppet-swift","branch":"master","topic":"rsync","hashtags":[],"change_id":"I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e","subject":"Support to configure rsync max_connections and not use xinetd for rsync","status":"MERGED","created":"2020-05-25 23:50:52.000000000","updated":"2020-05-27 19:58:32.000000000","submitted":"2020-05-27 19:56:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"730688-1590609404409-2a4b48c7","meta_rev_id":"4e11181d74211b436f08757e923d52e8d52946fb","_number":730688,"virtual_id_number":730688,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-05-26 23:41:58.000000000","_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-27 19:56:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":2,"date":"2020-05-27 15:09:08.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-05-27 17:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-05-27 17:23:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-27 15:09:08.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-05-27 17:23:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-27 19:56:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00d9d4807430f7aeb0f47b42648e13f7108ba959","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-25 23:50:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b01a4b9ec8a21c4fe1d245bc5aca4798f5fbb32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 01:17:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ddd75c35fe4d43b881b1d846a82a7f30 : FAILURE in 5m 26s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9f4fc827627849cb8a35a56f601b1c63 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca5469bd194542c9a236412988ea4ca7 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9426fd208f0841c295d3646379f6f80a : SUCCESS in 19m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/0a8274eb6a224489b3c592a2fed93f7e : SUCCESS in 20m 09s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/982c58d3f2aa43ac80cdb55e15fa4880 : SUCCESS in 18m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbb4130dcdf24b309b409856c70ecc3e : SUCCESS in 19m 46s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e81da6e7d85d4d1a8955313b5aec17a0 : FAILURE in 46m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/84a9494a0c1341bd966c291434821138 : SUCCESS in 58m 24s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a23406fb67f544fd89c62733f6b1a14c : SUCCESS in 31m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/2a97c9c33d6f402fa8090c21c81bc09c : FAILURE in 9m 58s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/db3d728e5d764862904c896b424d4051 : FAILURE in 6m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28f9efbc17cf4ff9888899f890993616 : SUCCESS in 3m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/df9961a32c474af08740ae295b2a6fba : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":1},{"id":"0c4cac912b5a6326740e5418d955a890e8c3e0dd","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-26 08:32:50.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe82c367f3d1dd75b0061fd365c7ccc1f901a317","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-26 23:40:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"907a00536855da32ecd7e2e0cf87955d9eea9364","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-26 23:41:58.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nThanks, yes the default value should be changed back.","accounts_in_message":[],"_revision_number":2},{"id":"d9330846cd9f23daeb5169183f40247a4e86921d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 01:11:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/490e57933d6448849df437fcd43d2dde : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d8eb41da52be40b38610df3776628536 : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2acd828eec464977add4990295e85c6f : SUCCESS in 4m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/492678aad66146ff92076caa1cdb6dee : SUCCESS in 15m 23s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c72c75fb3f7742e58cb31430b3cf2be5 : SUCCESS in 18m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a41bd57b17c5408898900527716438c2 : SUCCESS in 16m 07s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/641dc6b4fee64563b5afdd84b9d5fa1f : SUCCESS in 18m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9eeaf64723d464faa09bdc9f7e6dc39 : FAILURE in 34m 11s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6eba05e095df4589a5932af9e48cbd95 : SUCCESS in 1h 09m 07s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/52499317d7044d64bd8495ccda72e5da : SUCCESS in 30m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/03ecc926a75f4bd6a32536690749a900 : FAILURE in 8m 51s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ab4a32cf2c3f44d583913c315d725f7e : FAILURE in 6m 38s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ff1e6219a5144d094755a4977a2545c : SUCCESS in 3m 21s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2d54dd954d734b7685844cea8313856e : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b93fd6d5c109258d810d67d18e9f91d7f119fdfc","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-05-27 15:09:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d14639ef176c1e176c18ed8d240c63aef015834b","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-27 17:23:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ec173a623b0354dbfb5ebf3ad0d9ec8229af8977","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 17:24:03.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba22c9e57e2ca62e04d7f762acb62e8c7f6a693c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 19:56:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/234165d0b6a54e2c813cb7b423190d0c : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/41a58e6463d24466bf362695a63094ae : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c5bfbcd5c8f4319b7f943502689d3bd : SUCCESS in 6m 17s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/62914ab0d67d4762b2bcddaf91797852 : SUCCESS in 21m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/83084a6adf274acd98b6598d06e2f7d7 : SUCCESS in 16m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ebc99572d42547b388a0ad2a71ccc2cf : SUCCESS in 17m 40s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7ed30ea9e624beb979a72a4ea000fd7 : FAILURE in 42m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2a774e818f334dc9acf1bf368d38524a : SUCCESS in 1h 09m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/79a6f11b87604eafbd1a4ec8c0bcbe79 : SUCCESS in 29m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab5f30311fa4014b323c8dbda5de329 : SUCCESS in 4m 08s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4a2557d391e84a40a51b749ab7638c3c : SUCCESS in 1h 49m 08s","accounts_in_message":[],"_revision_number":2},{"id":"60792374edaa521d3a4846156d00a2a6dbe728e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 19:56:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"a2387671093bd029a0292d0540bb5a38900c6a68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 19:58:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5972e21d981e40e185025a30d84117b9 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237","revisions":{"63ccf7479f25001e54eb6d42b117abf897fd6881":{"kind":"REWORK","_number":1,"created":"2020-05-25 23:50:52.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/88/730688/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/88/730688/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/1"}}},"commit":{"parents":[{"commit":"f847dae469f14dfbcb881e2d18102e2bf1b8641f","subject":"Merge \"Fix disk mount options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f847dae469f14dfbcb881e2d18102e2bf1b8641f"}]}],"author":{"name":"Shi Yan","email":"shi.yan@unimelb.edu.au","date":"2020-05-18 15:22:25.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-25 23:45:35.000000000","tz":600},"subject":"Support to configure rsync max_connections and not use xinetd for rsync","message":"Support to configure rsync max_connections and not use xinetd for rsync\n\nChange-Id: I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/63ccf7479f25001e54eb6d42b117abf897fd6881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/63ccf7479f25001e54eb6d42b117abf897fd6881"}]},"branch":"refs/heads/master"},"ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237":{"kind":"REWORK","_number":2,"created":"2020-05-26 23:40:07.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/88/730688/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/88/730688/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/88/730688/2"}}},"commit":{"parents":[{"commit":"f847dae469f14dfbcb881e2d18102e2bf1b8641f","subject":"Merge \"Fix disk mount options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f847dae469f14dfbcb881e2d18102e2bf1b8641f"}]}],"author":{"name":"Shi Yan","email":"shi.yan@unimelb.edu.au","date":"2020-05-18 15:22:25.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-26 23:38:55.000000000","tz":600},"subject":"Support to configure rsync max_connections and not use xinetd for rsync","message":"Support to configure rsync max_connections and not use xinetd for rsync\n\nChange-Id: I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
