)]}'
{"id":"openstack%2Fpuppet-swift~731436","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"rsync","hashtags":[],"change_id":"I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e","subject":"Support to configure rsync max_connections and not use xinetd for rsync","status":"MERGED","created":"2020-05-28 05:22:13.000000000","updated":"2020-05-28 09:39:13.000000000","submitted":"2020-05-28 09:37:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731436-1590658652790-06738b85","meta_rev_id":"fec15fe9f7dd8130d209b7f7987d348c8f96cdf0","_number":731436,"virtual_id_number":731436,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 09:37:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2020-05-28 07:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-05-28 07:20:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2020-05-28 07:50:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 07:20:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-05-28 07:50:22.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-05-28 09:37:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae716b92bf503c626d70a54ddb414aab06f92fce","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-28 05:22:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7df3e49f1ccd3783d0ad17ddf2f3a5ac3d12b158","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2020-05-28 05:22:40.000000000","message":"Topic changed from ussuri to rsync","accounts_in_message":[],"_revision_number":1},{"id":"a7dab63c352d9d27178ec0bdffa833ccfd583568","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 06:51:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52d682978fe34675bd5205e68fb4f0fb : SUCCESS in 3m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/622529c7a8a0467abe8d91fb08bbba45 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0c53fbfdfcc4ec1bd8296f9cea1e03e : SUCCESS in 4m 01s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/be7eaa1525a04d779a967c1e9d0091bf : SUCCESS in 15m 07s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/8be5be96015241d1b1f7a4d619855295 : SUCCESS in 14m 58s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/150d9fc70dc54eab98bbbf9bfa7176d7 : SUCCESS in 13m 03s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/abf8f1074efe4c02ad14cc1cfc78253d : SUCCESS in 17m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbc713aa4d5042a38c2900a9fe315f52 : SUCCESS in 49m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/236297dd20124be696b8ba9251be2cf5 : SUCCESS in 1h 03m 10s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2cfb2d45c19442398f0c101ab21a18b0 : SUCCESS in 26m 08s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/dc32a224787e46a99031a4fd76e1bbb3 : FAILURE in 7m 08s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c48b8c94813d462da4305b126790a0a5 : FAILURE in 5m 40s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2fb298f417324b4eafa222a8503bcaa3 : SUCCESS in 3m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/35c8cd1ebc5943a79c1d5934979b6d09 : SUCCESS in 1h 27m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2e063faedf68a7c84e83e086d9941187fc4d44a","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-05-28 07:20:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"27130354940955f2c719fcb331dbac9dd53d0da0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-05-28 07:50:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6efaabfad49916f3664da37a4ddbdc32ffeb2575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 07:51:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"15e7bd68b940d648b85d5261e814884cb671e24e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:37:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d3501465fcd472805b0853d4abc4f5b1e1a691bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:37:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/522bfcc7914b46ae95c49351caa99af2 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce10b2505d9d4666977726430d4fae3d : SUCCESS in 5m 24s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c9833d9b66346f583ebfd01e1565b8e : SUCCESS in 5m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fed6e93da3424c31a4961bd8882a6fc3 : SUCCESS in 15m 24s\n- puppet-openstack-unit-6.11-centos-8 https://zuul.opendev.org/t/openstack/build/a28954ab98a7490c9280e97efae701c5 : SUCCESS in 15m 59s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a0a58fbc2ef4e4998fbd934ff426ef3 : SUCCESS in 17m 07s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaee995dae5d4efa963e42fa7f3aae6f : SUCCESS in 54m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b6bbbbefdadb4453ab3c8ae1e53a96a6 : SUCCESS in 1h 17m 22s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d36a88059eb14c0990c6173cab136c3c : SUCCESS in 28m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4902e3d706fb401fb418429feb0a9dfc : SUCCESS in 3m 30s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aa509b43d9d14adaa5685ad50535dd8c : SUCCESS in 1h 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fec15fe9f7dd8130d209b7f7987d348c8f96cdf0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:39:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cdbc1c27be824db09ef4bad5f42f4b82 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"241c22aff9e64030e40d0ba6cf4bc531617abf18","revisions":{"241c22aff9e64030e40d0ba6cf4bc531617abf18":{"kind":"REWORK","_number":1,"created":"2020-05-28 05:22:13.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/36/731436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/36/731436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/731436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/731436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/36/731436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/36/731436/1"}}},"commit":{"parents":[{"commit":"b88138c32c76aeee9703a31142814301e2a4ea93","subject":"Merge \"Passwords should be secret\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b88138c32c76aeee9703a31142814301e2a4ea93"}]}],"author":{"name":"Shi Yan","email":"shi.yan@unimelb.edu.au","date":"2020-05-18 15:22:25.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-28 05:20:59.000000000","tz":600},"subject":"Support to configure rsync max_connections and not use xinetd for rsync","message":"Support to configure rsync max_connections and not use xinetd for rsync\n\nChange-Id: I40f4e7f2c81cebf8213057a341ef81f7f3d09d1e\n(cherry picked from commit ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/241c22aff9e64030e40d0ba6cf4bc531617abf18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/241c22aff9e64030e40d0ba6cf4bc531617abf18"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
