)]}'
{"id":"openstack%2Fpuppet-swift~734126","triplet_id":"openstack%2Fpuppet-swift~master~I123ab64e77076d2eada6c9cdc4f6526398b071d1","project":"openstack/puppet-swift","branch":"master","topic":"signing_dir","hashtags":[],"change_id":"I123ab64e77076d2eada6c9cdc4f6526398b071d1","subject":"Deprecate the remaining option for Keystone PKI token","status":"MERGED","created":"2020-06-08 15:27:33.000000000","updated":"2020-06-11 16:37:56.000000000","submitted":"2020-06-11 16:34:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734126-1591893286854-d9d06e61","meta_rev_id":"b2fcd70502809f649f0b082aea3e479b6201940f","_number":734126,"virtual_id_number":734126,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 16:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-06-10 18:23:28.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-06-11 14:40:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2020-06-11 14:40:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 18:23:28.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-06-11 14:40:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-06-11 16:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e0f439f9e4ffa5f013888e3116ba1ed13cfaeb7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-08 15:27:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99fb2da845e31a4cec281bb312c63d159b5c4a89","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-06-08 15:41:03.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"aebdbe69eadc1216a2c874dd95f22338440388d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 17:10:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c8360a54d64407180528ca72f175651 : SUCCESS in 5m 10s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/da7f42e05ef14a218f908d7964b853d8 : SUCCESS in 6m 16s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66ef7d01b0c34076a8abc1e08736547d : SUCCESS in 6m 29s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/56c4c20887504da69e1263f4867c1feb : SUCCESS in 18m 57s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2bd553f7574249fda130be5fa4ae559f : SUCCESS in 22m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e557584e92384d26ba29f51ddb8f456e : SUCCESS in 17m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d6ce93e39e34fd4993381004aa8d323 : SUCCESS in 21m 19s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a84b6541ea364e3a8dc13fdf49bd18b6 : SUCCESS in 55m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/8fe856a79746495bba0914a958b79f02 : SUCCESS in 1h 03m 31s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/94b4e62e60fc424788fcb8e2169d487c : SUCCESS in 35m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/882974b57737478d8dbb8bea74460899 : FAILURE in 12m 39s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdb3f5f39b21458dad278a4ef098b0bb : FAILURE in 8m 34s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c2fac241b344c08a9ecc5898a3aa7dc : SUCCESS in 7m 14s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/52fbefeaaa99479bada66791393e17b9 : SUCCESS in 1h 25m 39s","accounts_in_message":[],"_revision_number":2},{"id":"11d789a34c1ec758a64f3bc743103629ae75111b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-06-10 18:23:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"adef6f8c7d73a23e2f116ee787e5e094932daa17","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-06-11 14:40:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"bcd1a15c8bf096c2a60d8a06865691a894d098ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 14:40:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5444d3d137da6f93e03d1d2e7fcf6be2ce59c219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 16:34:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f397c600ba545068a6486bdb3340b67 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6617607ae39484b882058895235a09e : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e13500118b94254ad8165fe5418e302 : SUCCESS in 4m 54s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/777a4643bff04c9cb089399eea9467c2 : SUCCESS in 19m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e7bfc913f0e040b7bf1d5b6bb20ab192 : SUCCESS in 21m 58s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0b6e162732f473c822a4f53ddc718dc : SUCCESS in 18m 26s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5609e575b6dc45eaa7bb6bf4c4331a7c : SUCCESS in 1h 01m 07s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/bdc08517d978404bbad4a3dcdfd08eda : SUCCESS in 1h 11m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/4d22b0934b6943b09d3f48447378b195 : SUCCESS in 28m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/09287276b61b43a5bb09e18b230ef402 : SUCCESS in 4m 57s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ea08cec1420745a5b73664d316c9fd5e : SUCCESS in 1h 52m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f5f74eccfffd40b5abbc2f02b885be832a8f3f18","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 16:34:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cdd28ba712d1f3fe20e8144507d93e993865fe55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 16:37:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf494c7d46734790ba2a0196685b5420 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"424b861a57d60cefeeebc2fafaed97484fac261c","revisions":{"1f13fad121856bbff83bbb7339d738ccbd8f51f1":{"kind":"REWORK","_number":1,"created":"2020-06-08 15:27:33.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/734126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/734126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/1"}}},"commit":{"parents":[{"commit":"ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237","subject":"Support to configure rsync max_connections and not use xinetd for rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-08 15:26:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-08 15:26:16.000000000","tz":540},"subject":"Deprecate the remaining option for Keystone PKI token","message":"Deprecate the remaining option for Keystone PKI token\n\n... because it was already removed from keystonemiddleware.\n\n[1] b3e84aafc0302b1a87754f438696794076ba844f\n\nChange-Id: I123ab64e77076d2eada6c9cdc4f6526398b071d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1f13fad121856bbff83bbb7339d738ccbd8f51f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1f13fad121856bbff83bbb7339d738ccbd8f51f1"}]},"branch":"refs/heads/master"},"424b861a57d60cefeeebc2fafaed97484fac261c":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-08 15:41:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/26/734126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/26/734126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/26/734126/2"}}},"commit":{"parents":[{"commit":"ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237","subject":"Support to configure rsync max_connections and not use xinetd for rsync","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ff9608a8c100b0acff7cdf6eeeadcd3d68c4f237"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-08 15:26:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-06-08 15:41:00.000000000","tz":0},"subject":"Deprecate the remaining option for Keystone PKI token","message":"Deprecate the remaining option for Keystone PKI token\n\n... because it was already removed from keystonemiddleware[1].\n\n[1] b3e84aafc0302b1a87754f438696794076ba844f\n\nChange-Id: I123ab64e77076d2eada6c9cdc4f6526398b071d1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/424b861a57d60cefeeebc2fafaed97484fac261c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/424b861a57d60cefeeebc2fafaed97484fac261c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
