)]}'
{"id":"openstack%2Fpuppet-swift~741352","triplet_id":"openstack%2Fpuppet-swift~master~Ia0cc5d8aa3e4daaf2be0889f82d2186e92553348","project":"openstack/puppet-swift","branch":"master","topic":"relative-name-inclusion","hashtags":[],"change_id":"Ia0cc5d8aa3e4daaf2be0889f82d2186e92553348","subject":"Eliminate remaining class inclusions by absolute names","status":"MERGED","created":"2020-07-16 01:26:45.000000000","updated":"2020-07-27 03:52:00.000000000","submitted":"2020-07-27 03:52:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"741352-1595821920524-bdae804c","meta_rev_id":"6d37869a990e5c9aa20b742809116b7cfa9b920c","_number":741352,"virtual_id_number":741352,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-07-27 00:00:33.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-27 03:52:00.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-07-16 14:14:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2020-07-16 08:34:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-07-16 14:14:20.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-16 08:34:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2020-07-16 14:14:20.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-07-27 03:52:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b3cb1e7721f5cf1bcdcfef76522c76b4c5d77e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 01:26:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48e4c3aa6bd335a4767f5ba937fefdf86738b1da","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 01:27:44.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f990117a670a561c8c0c79ff5e402d19dd240c2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 03:08:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f077890a46ae4cd3bda38d5f29a04d17 : SUCCESS in 6m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80a448c2e67848c0ab0274bf7947585d : SUCCESS in 4m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07ebcab6bded4fdc9b337cf8a4654c28 : SUCCESS in 6m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/bd8d7a7631974d32ae8a1acfb2c705b0 : SUCCESS in 15m 42s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/dd00ab0f82614192994982549580b175 : SUCCESS in 17m 46s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25d1de64f9ac4c488410ce897e031fe1 : SUCCESS in 16m 33s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d48f05af0cb488d8ae95c83382f2984 : FAILURE in 8m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ea4ab3f3f1e14fd5b49806141540b658 : FAILURE in 55m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/43cf6d64c36e45939d672d9253e5528f : SUCCESS in 1h 32m 33s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/99c8688a883f4d29ba48371f4216b306 : SUCCESS in 26m 59s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/459cc4d49d6d44f9acda772ddff11508 : FAILURE in 10m 09s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/282300755eaf49d5a79c672731592c36 : FAILURE in 6m 22s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3abe1d8b151c43bfb39a5e258c9676df : SUCCESS in 1h 39m 14s","accounts_in_message":[],"_revision_number":2},{"id":"8950200a9a8f613e8bc8acf979b9dc13c9264a35","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-07-16 08:34:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"489e2cfc7cbce8804b3eb7dd19835e091855cf3d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-07-16 14:14:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6cf33f45b4038789c883c6bc91d4e154aac9c0be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 14:14:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc79f3364516a76b9bbb0644e10a344b3a270655","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 16:06:37.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5321201214394e5e870cdd94e3940d3b : SUCCESS in 5m 48s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e32c3cd2cad42c3b02486b80dafd2bb : SUCCESS in 5m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47a145d48acc430eb3982ff5cf898440 : SUCCESS in 5m 20s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/6aab6c70467d49c69430c5ad12ef3c26 : SUCCESS in 17m 13s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9e36be27b63f45bf8b886f261e20bd40 : SUCCESS in 20m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e73f7b44370c4e138c8441ce9414f987 : SUCCESS in 25m 47s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c58493817d047199a3e4b255c407412 : FAILURE in 49m 18s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5429da947db34eec84574049b72ea126 : SUCCESS in 1h 28m 45s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/612dd3099ed741a5a15697f6657670cd : FAILURE in 9m 13s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e41fa071e5954d9dbc011e49f8363407 : SUCCESS in 1h 48m 08s","accounts_in_message":[],"_revision_number":2},{"id":"fd00874f781e3b727866282eb813608f3de0216e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-16 23:03:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b753dcfad67db8862d6b5b6ca8010617f477182a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 01:12:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3d7b5c987534662b1128b2c794b581f : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/823df172cfdc4a849c2d146524f2188b : SUCCESS in 5m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64d35a7b1fa1443c9336df56ba753e3c : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7cf457fc8db54fb38efd0fce85ce36e2 : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/122f9fe815844301b81ce0d38a58e55d : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59418345129b43f38008d5cb0d791cc7 : SUCCESS in 17m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42a4379e37df4edea6d22556ce11ec12 : FAILURE in 7m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/055617656e52466bb4c8514ad3f6f44c : FAILURE in 49m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6b9c24e8413b486882b55869e6900399 : TIMED_OUT in 2h 01m 00s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6fedc95f298b41a1b0de04328d9b9bef : SUCCESS in 27m 33s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/03b8eab1c8384c7abe5239cf09ab7c85 : FAILURE in 9m 48s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efb0b5dc1ffa4acfa2ad1377c50268c1 : FAILURE in 6m 58s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1bfd143632a6455892c3bfca07fcc666 : SUCCESS in 1h 40m 36s","accounts_in_message":[],"_revision_number":2},{"id":"7b93a660c0a2700da9d24791530d474dacd05756","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-17 05:03:52.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"db61109736a03619ae0a83b8e632d55be2fde735","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-17 06:38:27.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c21298cb77f6426a8e1074501adbdfa2 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a5c7d55bb2dc4765b25a8f717c7e2fe9 : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9820407ebcac41c3b1c4a2d97a48ee57 : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d2eff0df17914e4c81c9be7299afa084 : SUCCESS in 16m 58s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/66301fe9d460462b8c15a5bf10168b28 : SUCCESS in 18m 19s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/34f8ff76e3034043b0fb17212ba9d00a : SUCCESS in 18m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c90f28931d74a33afed67308e25479f : FAILURE in 8m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ef7f78681a5e4b6293a4da26a4e3e7c1 : FAILURE in 50m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cce8439f458642eab2710cb986a5c72f : SUCCESS in 55m 21s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c4946ad502d84b66878cb07c31ff0816 : FAILURE in 10m 35s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e58a3bc71ffe4bc1a00dde75868d5ed8 : FAILURE in 8m 37s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eadbf2a5d7594bbcb6eb7388339fbeca : FAILURE in 6m 45s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/27aa356e38c3459191d681a6de15dfd2 : SUCCESS in 1h 22m 30s","accounts_in_message":[],"_revision_number":2},{"id":"0ecf9bc365d0e4882907792922e5f42c6f20a1bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-18 03:27:02.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"caf8f0db411aa025c8e71e4d9006fd39e7189ca9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-18 04:51:57.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0aaf6b6575bd434684b3221833067a88 : SUCCESS in 4m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65e8e405dacb4f78b041c4b34522a2e2 : SUCCESS in 3m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f590a7635a05497993ea0f183c388617 : SUCCESS in 3m 50s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/aefe2daf954a4bc38148c9c1f3fb9323 : SUCCESS in 16m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5d4fee8e2a4c49f692296c40836b0de0 : SUCCESS in 18m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f9ea4721d384001a21becb130663211 : SUCCESS in 15m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67467d4136974329b602ea4e4d296153 : FAILURE in 7m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e55530f3e2240dea9214f3c3dd1756f : FAILURE in 49m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e2174c39b7ca415b92491f270864e6d5 : SUCCESS in 1h 23m 18s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/178d1a9b7300406fb1005d98f3362a55 : FAILURE in 7m 08s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/3650ca6c72e448de84319549ea2fd661 : FAILURE in 10m 35s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f2fe9ebd13a4d8cae7cedffeaccbfe3 : FAILURE in 5m 26s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2504970ee37446c0bc3cf178a839c9a2 : FAILURE in 53m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6ea77ee8960680a0c31a6106d85ea93de0f0ae69","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-18 23:26:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"630e37761f59c2103a2ca2632220fc601bfdcf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-19 01:17:42.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09ec6bee7bdf49368ee4f7a88c6e7160 : SUCCESS in 4m 19s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/392159531c7a4ab4933e74eae161fe41 : SUCCESS in 3m 37s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/49d2f699e4b746999f430a2e28080794 : SUCCESS in 4m 31s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/21359bcf8c894daa9295a1953cd5558a : SUCCESS in 15m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f4e03d690bae439da18cfd98ac3553b5 : SUCCESS in 16m 43s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6b3390c3f7c4b6e97acfb4dc3745c46 : SUCCESS in 14m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/880f3614a139438b8b37156dbac468f8 : FAILURE in 7m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3ec963074f24ab1bf98f5139c116c7e : FAILURE in 49m 34s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/072699266a5548d7b581824586eecef6 : SUCCESS in 1h 29m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/9d11fdc3f59344fc8ce9bbad84d54659 : FAILURE in 8m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/fe94383efe404e98949c78bc13ac34d8 : FAILURE in 9m 24s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9355dec90ff4389a15a2b63fb126b4c : FAILURE in 7m 03s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1902b01abe75428387e769275145c4b4 : SUCCESS in 1h 49m 45s","accounts_in_message":[],"_revision_number":2},{"id":"a9b0d4d19e277d33d9e1a8e167d5615531462d8a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-19 22:48:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"af42a55b502e90493ebd7541e09a901cded7a2a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 00:31:25.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203efc3a20e84b8eae0ecd0a7f9dc4f0 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54436afbe21f42ef8b78a21107a10574 : SUCCESS in 3m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/281c0628ee9b469ea0a934283cd6851f : SUCCESS in 6m 24s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/9cea6643ffad4b0381c06871d007a5da : SUCCESS in 18m 28s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/14c96b32fe8e4707aa5846dda7955aef : SUCCESS in 18m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/805cee2ff8bf4e1383b4c7171678d575 : SUCCESS in 16m 01s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8adc34ea5db94765a67ac2820aef65f1 : FAILURE in 6m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ebfa80ab3624b69886a16eee92aca3b : FAILURE in 51m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/b743db78ab7045778bd8102d92e61d14 : SUCCESS in 1h 31m 02s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/ddea6589e2434cff8be35a768dc55eda : FAILURE in 8m 29s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c8c8604f17fe455dbcf82f2fe74bb3bd : FAILURE in 8m 17s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/780ef744565f4d5f88f47116ba761f8f : FAILURE in 5m 48s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e29ad20630bd4eef8ead261e34a31f30 : SUCCESS in 1h 37m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5db38dbe215f27b2ef9dcf3621315c58b2eeda36","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-20 00:34:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5c99f7883faae96425a2891d43497e5017c1473d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-20 02:13:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8461ca9bf644179ae421714046cdb83 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10a341a80d764f10ae630339fef2916a : SUCCESS in 4m 10s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59273bc3e4ed4ccc916ff8e267f4b67d : SUCCESS in 3m 22s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/ffcc1a66e87a4569b3ddbabe82801d04 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/574c1a0c3b804c978c9fe0bfb66e217d : SUCCESS in 16m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b24142ea2734893b8012774ff314c71 : SUCCESS in 10m 43s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/271ee4f3929b4130835d1e1d0c1a5d2c : FAILURE in 5m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/731e217781f64ceb9730680353cea976 : FAILURE in 50m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/713d7ed291634ef9912bd1e666ab471e : SUCCESS in 53m 19s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/08a2fee0661d4b3babec3c82af762ff1 : FAILURE in 8m 45s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e47c43c74c8945b683d9a90d64e7e247 : FAILURE in 7m 15s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e5cab6e7cdc4b30b7fdc1c02a8c80a5 : FAILURE in 5m 40s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ac184f8bbd654f91bc525c79e23ce74d : SUCCESS in 1h 37m 45s","accounts_in_message":[],"_revision_number":2},{"id":"12e430b5913c43aad170c5dd3ddfb3ab1d70f430","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-21 10:06:43.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"36354f7c642d3b2780bbab1732edae31cd7fe8a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 11:58:05.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66d083b1bbda4413813c16c01650d51d : SUCCESS in 5m 17s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d21526e4d9442079f6af8a3289f06ee : SUCCESS in 5m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/678b1703be3949858d69eea1b9f18431 : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/d7b03e370cfa4e0a9d9c2030be17592c : SUCCESS in 22m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/cd6cf9257b3b49778fc7e2e759251ff9 : SUCCESS in 20m 48s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b4a688bb62145b894b5e60fbb11264a : SUCCESS in 17m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5025a9941914480dbcbb7bc9e8df9663 : FAILURE in 9m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdd572d9981d408494983e9fced65ff3 : FAILURE in 50m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/6331a69ca07f41e1bd0d201f656a1781 : SUCCESS in 1h 24m 47s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bb2e27cb9faf4386ad209b49de86d3df : FAILURE in 10m 09s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/bbddaa2fbf18404a956078db5e3726c7 : FAILURE in 9m 19s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdb978bb57374c9d8f55e18b27dd2379 : FAILURE in 8m 17s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f2cb42fd14954a15bdffaca4ee1598b8 : SUCCESS in 1h 44m 11s","accounts_in_message":[],"_revision_number":2},{"id":"b4c41a1dcd96f4217be186e7085f478f8dbaaa78","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-23 14:48:38.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2c794d25153c8df628d8a4c3d5c0fd7f2cae2e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 16:55:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75be40080bd14085ba27fee432562f41 : SUCCESS in 6m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c891b8a4016e437cb93e07736d31e1b0 : SUCCESS in 7m 12s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2578f2b1d8f64e86b50adc03e3f606fc : SUCCESS in 4m 49s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/320e6ad2c7884e638e5788cb67ef85a0 : SUCCESS in 17m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f46fc9fd77b34e24b1603a321a7cff55 : SUCCESS in 19m 47s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ac5d056d92a47d0b7feb7ccd65d0a07 : SUCCESS in 17m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30fe76b7d354493ba3784fa5e9ffe85d : FAILURE in 9m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06526aa30ab94c319f6db253c969956b : FAILURE in 1h 18m 32s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a3f1911996424a57961d5f2dcde1d502 : SUCCESS in 1h 56m 25s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/25d00b2390244988bb46b9049f7279e2 : SUCCESS in 35m 00s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c8676eab39dc48978ce10806c417f084 : FAILURE in 8m 45s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52cc8a1c447942168d351fd1fd614140 : FAILURE in 11m 26s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/4a3eb10fd9dc4c6387c82d44e3f32c3e : SUCCESS in 1h 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"890953b661793c0033b7de942787ae83d57479f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 16:55:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68595adf6cc3d7cc8b630489547f4c94240be1b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:07:34.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64d60620c3864a26910923f6a413130a : SUCCESS in 5m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3334c7d091b34ca1be267705397c8b7b : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c20f0e97534d4783bd89d2cf1f78d3f4 : SUCCESS in 7m 03s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/fe4f5f8debb34b1c8d3f71f71beaf64e : SUCCESS in 16m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/8f5f229fa3794e6ba66a90c070c86e90 : SUCCESS in 21m 02s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b63fea4952d649e7a963433f9db6fd9e : SUCCESS in 12m 58s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4bd3747eff81421da4184b0a886a8871 : FAILURE in 36m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/752348e7d6194035a60667d3025dba83 : TIMED_OUT in 2h 01m 53s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6dc02bcb8c234003803f20ff4d3fd780 : SUCCESS in 19m 56s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d303b996bec345898b8bd0acb056b6c9 : SUCCESS in 1h 49m 50s","accounts_in_message":[],"_revision_number":2},{"id":"432937cd249108c2e15fe95a88c0d9a46d65bbf7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 00:00:33.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d4a03d2eed052e925354bcbe79221ced0b7ceecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 01:52:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77ce8f0406df43a0813f88eb6321f3a3 : SUCCESS in 4m 41s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e488672aa234187aa333bc3d70046fa : SUCCESS in 4m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c01fcd1c837443db143782dc6ab547a : SUCCESS in 7m 15s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/46998ea38148458f85260b1b216ccb49 : SUCCESS in 16m 03s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d7c36e50309849439f21f5f78975061a : SUCCESS in 13m 25s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8326646b84f49c584b9de0b20de7278 : SUCCESS in 14m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b42342f45c604bb99dd6a2f896da81d0 : FAILURE in 7m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99242d75701c491abfbb54c1bb652990 : FAILURE in 48m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/fa85736da4d74ccb8c50af8c3e72f200 : SUCCESS in 1h 28m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/b9cf8b6454c74440b489a21de8e65a7a : SUCCESS in 28m 10s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/c65623e0e14148c4b08348f52b430d2a : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7dd50834b4b4bdbbb426a48be28dc97 : FAILURE in 6m 17s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/071dec320bd04a478578bda63123dd77 : SUCCESS in 1h 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"e6f4bdf7f806e1cb01930df4c543a2b8b44478a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 01:53:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"740c188620c28d67f2565a94026d903498b382ea","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 03:52:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3a080a49b2b799e8433e6da67c2745abb6c17dc5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-27 03:52:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/80c2ab805f6948cca13e020162e5931e : SUCCESS in 5m 12s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d346867492b54877be188daa5534f8bc : SUCCESS in 4m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a7742dce3f845119de70711c6e8d707 : SUCCESS in 3m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/aaf46466845e4e6483bb30af9e254473 : SUCCESS in 15m 50s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/46ea79d569224f659cc939f4bc3f0acb : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3e0391b70914054bcd9b12cb6a67ae0 : SUCCESS in 16m 55s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ca43542b40c4f569ec91bd0f26fae6a : FAILURE in 51m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/58a95eb4d1194539bc032628501c227f : SUCCESS in 1h 52m 17s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/7af9961d2c0e4710b1ebe44c73f92ba4 : SUCCESS in 28m 27s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6595a1d7c1c44c3fba28bb5c07d9912a : SUCCESS in 1h 34m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7a6fa33e77406a7fcc0c28b71eff1fb9b8f1a115","revisions":{"c6febe3667707967ca9823b04cf591d167a6d1cf":{"kind":"REWORK","_number":1,"created":"2020-07-16 01:26:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/741352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/52/741352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/1"}}},"commit":{"parents":[{"commit":"c3e8ab9bc04edd38452f5e64cfb64510917ca27d","subject":"Use relative name for class inclusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c3e8ab9bc04edd38452f5e64cfb64510917ca27d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-16 01:26:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-16 01:26:03.000000000","tz":540},"subject":"Eeliminate remaining class inclusions by absolute names","message":"Eeliminate remaining class inclusions by absolute names\n\nChange-Id: Ia0cc5d8aa3e4daaf2be0889f82d2186e92553348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c6febe3667707967ca9823b04cf591d167a6d1cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c6febe3667707967ca9823b04cf591d167a6d1cf"}]},"branch":"refs/heads/master"},"7a6fa33e77406a7fcc0c28b71eff1fb9b8f1a115":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-16 01:27:44.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/52/741352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/52/741352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/52/741352/2"}}},"commit":{"parents":[{"commit":"c3e8ab9bc04edd38452f5e64cfb64510917ca27d","subject":"Use relative name for class inclusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c3e8ab9bc04edd38452f5e64cfb64510917ca27d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-16 01:26:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-16 01:27:34.000000000","tz":540},"subject":"Eliminate remaining class inclusions by absolute names","message":"Eliminate remaining class inclusions by absolute names\n\nChange-Id: Ia0cc5d8aa3e4daaf2be0889f82d2186e92553348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6fa33e77406a7fcc0c28b71eff1fb9b8f1a115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7a6fa33e77406a7fcc0c28b71eff1fb9b8f1a115"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
