)]}'
{"id":"openstack%2Fpuppet-swift~742163","triplet_id":"openstack%2Fpuppet-swift~master~I83ac4bea1d7282c48e87535000ec040668c5d477","project":"openstack/puppet-swift","branch":"master","topic":"test-beaker","hashtags":[],"change_id":"I83ac4bea1d7282c48e87535000ec040668c5d477","subject":"DNM: Test Beaker job","status":"ABANDONED","created":"2020-07-21 11:53:27.000000000","updated":"2020-07-27 00:11:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9b5e604f4abf48e549b197aa850d87bbd7087008","_number":742163,"virtual_id_number":742163,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-07-21 14:03:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 14:03:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"657519e685dd1a4601f5007ba79249d49beba6ba","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-21 11:53:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3adecca035ebeca7edf42c008ae6f37e043a17e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:03:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4832296972a3403f9d08ca7cc8efe8d2 : FAILURE in 5m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afe211170f7c4232a7589ac6e47ff21d : SUCCESS in 5m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a386e841c854bfca78c705ef1ff2bf7 : SUCCESS in 6m 28s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e54b8750a3e34118ac275b42f8cbf392 : SUCCESS in 19m 43s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/489fe5e4f7804629b28564075439a99d : SUCCESS in 20m 06s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9de7f56837ed44918a69a4aedda11f0c : SUCCESS in 19m 12s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ff5b31abc224050a86d10f2f556c424 : FAILURE in 9m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c9c49d5fc5f4ec589798da604c48936 : FAILURE in 50m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/63b1e54a67194c2d8a8c3afbb9ec3315 : TIMED_OUT in 2h 00m 54s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/11e4617c1a8d4ad59710749667c3f0e8 : SUCCESS in 28m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/5f7b9bb31eb34bc1a048cb54349929d0 : FAILURE in 10m 08s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8ea3dbad1f5f4401bc6ea2db30bdace3 : FAILURE in 8m 49s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e88632227924b3f9d73bc091e841c0f : SUCCESS in 1h 43m 34s","accounts_in_message":[],"_revision_number":1},{"id":"282731a6abab2f75173678bb600256fbb946f729","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-07-27 00:11:20.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"288093a55edd63261d7316e81873970f6dd2be50","revisions":{"288093a55edd63261d7316e81873970f6dd2be50":{"kind":"REWORK","_number":1,"created":"2020-07-21 11:53:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/742163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/63/742163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/742163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/742163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/63/742163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/63/742163/1"}}},"commit":{"parents":[{"commit":"c3e8ab9bc04edd38452f5e64cfb64510917ca27d","subject":"Use relative name for class inclusion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c3e8ab9bc04edd38452f5e64cfb64510917ca27d"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-21 11:53:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-07-21 11:53:00.000000000","tz":540},"subject":"DNM: Test Beaker job","message":"DNM: Test Beaker job\n\nDepends-on: https://review.opendev.org/#/c/742146/\nChange-Id: I83ac4bea1d7282c48e87535000ec040668c5d477\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/288093a55edd63261d7316e81873970f6dd2be50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/288093a55edd63261d7316e81873970f6dd2be50"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
