)]}'
{"id":"openstack%2Fpuppet-swift~751278","triplet_id":"openstack%2Fpuppet-swift~master~I2f56bc456427d5deb30d86ad70e890d0897ef0c9","project":"openstack/puppet-swift","branch":"master","topic":"recoverable_node_timeout","hashtags":[],"change_id":"I2f56bc456427d5deb30d86ad70e890d0897ef0c9","subject":"Add proxy config option to set recoverable_node_timeout","status":"MERGED","created":"2020-09-11 13:36:44.000000000","updated":"2020-09-14 03:13:50.000000000","submitted":"2020-09-14 03:12:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"751278-1600053166225-c6eda660","meta_rev_id":"35227380fa22fc75eefb035aefed5e5b14b90970","_number":751278,"virtual_id_number":751278,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 03:12:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-09-14 01:24:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-09-11 19:40:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-09-14 01:24:59.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-11 19:40:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-09-14 01:24:59.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-09-14 03:12:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4db4900667ce662168f47158c1be30cd5f1db488","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2020-09-11 13:36:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f954f08193b4517fc66e0117f457dd47ea2353df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-11 15:05:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6536642b47fd41568472c0c37d19c054 : SUCCESS in 5m 49s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b1489e70d5534082813a04e671614cb1 : SUCCESS in 5m 10s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2041e0633dcc43aeb9a6ff1e761d5fd8 : SUCCESS in 17m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/13d1724da46740a6ab3adeb1ffb33cdd : SUCCESS in 12m 37s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a517add32c945aabfe640efdaa667d1 : SUCCESS in 20m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/160ed1f8abfe4f1c9167de08eea8ba39 : SUCCESS in 46m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e7026508d67c4277bbb2ad843a695b72 : SUCCESS in 1h 19m 35s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0dfe5260fe2742398a511a7d410180b7 : SUCCESS in 29m 58s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54f46f479b1f423084db45ab0e8e3835 : SUCCESS in 24m 10s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6fb39f520fa4769b2d5d290d09fff1b : SUCCESS in 4m 00s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/acd2948c1b84431c8a0694eeed87d0ef : SUCCESS in 1h 23m 15s","accounts_in_message":[],"_revision_number":1},{"id":"61726718a7b4a00aebe02b14466158e350d41ede","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-09-11 19:40:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2efc836e08587e3e581cffab26c612fa1c5a4f21","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-09-14 01:24:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f32654964c464d9955ca012d45f167a2c7015622","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 01:25:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c2722f701e04fec07df353846e3b800d83a62014","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 03:12:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75b7288dbd1d4483bf265153ea894850 : SUCCESS in 2m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af512a17f98647b7af8ae60143190bf7 : SUCCESS in 2m 55s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/4bee47c1f1b347a29ab432a887dd2a1d : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b3190e4f29048278bacfdccd6006cbe : SUCCESS in 11m 49s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7f2ae2eb94fe47a7af55da95dec323d1 : SUCCESS in 1h 04m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/199d7b3f83eb45a39f3f31e020530507 : SUCCESS in 1h 28m 08s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0d00bc328ba2429991e92c383013d2ad : SUCCESS in 30m 43s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0b83675716440508c62d0670958b438 : SUCCESS in 21m 50s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e444b81dc944706b9edba43c428424f : SUCCESS in 2m 53s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/1f367581bbbe4f97a38d9f9940fb6c70 : SUCCESS in 1h 42m 10s","accounts_in_message":[],"_revision_number":1},{"id":"89e00434256aed9750271cc6dbffb5be32221932","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 03:12:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2cc9274ff1e82eeb61b706b339d33a8f81be72ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 03:13:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/970fd3f0ef654c5088c50f6dfcc4f5a7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4d9315ec54eeb5878088cccf37ef12e8dc56ede5","revisions":{"4d9315ec54eeb5878088cccf37ef12e8dc56ede5":{"kind":"REWORK","_number":1,"created":"2020-09-11 13:36:44.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/78/751278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/78/751278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/751278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/751278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/78/751278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/78/751278/1"}}},"commit":{"parents":[{"commit":"0504cf2dbff1003ecd27744ce2555cda257effc2","subject":"Merge \"Add Puppet Litmus\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0504cf2dbff1003ecd27744ce2555cda257effc2"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2020-09-11 13:17:58.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2020-09-11 13:24:21.000000000","tz":120},"subject":"Add proxy config option to set recoverable_node_timeout","message":"Add proxy config option to set recoverable_node_timeout\n\nThis setting defaults to node_timeout if unset, which affects HEAD, GET\nand PUT requests. However, it makes sense to keep the timeout for GET\nand HEAD requests shorter to allow quicker recovery and prevent client\ntimeouts.\n\nChange-Id: I2f56bc456427d5deb30d86ad70e890d0897ef0c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4d9315ec54eeb5878088cccf37ef12e8dc56ede5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4d9315ec54eeb5878088cccf37ef12e8dc56ede5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
