)]}'
{"id":"openstack%2Fpuppet-swift~756835","triplet_id":"openstack%2Fpuppet-swift~stable%2Fvictoria~Id1566ae7f851707f2243b6da2eb8290563369e5e","project":"openstack/puppet-swift","branch":"stable/victoria","topic":"create-victoria","hashtags":[],"change_id":"Id1566ae7f851707f2243b6da2eb8290563369e5e","subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","status":"MERGED","created":"2020-10-08 14:47:18.000000000","updated":"2020-10-15 09:42:32.000000000","submitted":"2020-10-15 09:39:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"756835-1602754787151-6db76d6a","meta_rev_id":"eedd7d15a364902f3391c7e3298d2ea8b2023ea8","_number":756835,"virtual_id_number":756835,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"date":"2020-10-15 05:45:54.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-15 09:39:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-08 14:54:10.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2020-10-13 13:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-10-13 13:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-08 14:54:10.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-13 13:27:46.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2020-10-15 05:45:54.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2020-10-15 09:39:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9726c55eed58e10f402a49adf6deff2fdd9a9b16","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-10-08 14:47:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb215706a40ed830c710dc05e85cd59ed927b04b","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-08 14:54:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2074ac112ca2a0ec89b86deb79e9af0cf677850","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-08 17:46:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f13099279484b7181d25faa9065c074 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d27fe85e64e64826ad756187bf80a5bd : SUCCESS in 3m 36s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/9c3d57fab0fe4aca8b42a10aa1e891c0 : SUCCESS in 14m 22s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/997f847ced3940faa6cdb5eeb683650c : SUCCESS in 15m 04s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/32938993a2d445e0af2d2185e91090f5 : SUCCESS in 17m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7e6853ff5e1462895b366120aaec0ab : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/43f548bb5a76478c9a2db0c6ed46faa5 : SUCCESS in 1h 26m 48s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f22ddc25a580400e901d8c1b09d1cdca : FAILURE in 9m 13s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cfd0d28b423e490ca2a2a12a793bd1bc : FAILURE in 6m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e61fc915ef949958ff2d8d6b8239d77 : SUCCESS in 3m 32s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/744468caf6a540b9ad89fa1ec801a53d : FAILURE in 17m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d3a47cf39a1af5235315036c3a7d90c0d79277ce","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-13 13:27:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"37184b502e07ca0a93124e80eb23e03798371a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 16:21:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7af011178c7246968be2f89b4e3b6b99 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c10dcf4425a44379ab0c4ebc6c93ce25 : SUCCESS in 4m 29s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/c1db4c646e5a4d0bacd92e6c203dbac0 : SUCCESS in 18m 14s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/074ba87fb5b84666adaafd51d37afe09 : SUCCESS in 11m 30s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b092a57948504477a0967a567843c857 : SUCCESS in 15m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d9adbaf2cf64bb3b69fad895f0b4103 : SUCCESS in 49m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/433dbf5a7a1d4fa385a038f978a1b096 : TIMED_OUT in 2h 01m 08s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/170bef33a3f042cc8077a203ecf68705 : SUCCESS in 28m 27s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d7ff91570bc347c387d1bc164b983b0e : SUCCESS in 22m 02s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4746beb653c24c2bacaafb16881354d9 : SUCCESS in 4m 32s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/40ab2c57c56346ddb7cf2140092821d6 : SUCCESS in 1h 30m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9752c84cd054f79e2ed6ab1696f81db51123a3db","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-10-15 05:45:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c7230c55148e792527c75dea3e2da62bdb8df7a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 07:42:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90d63f39abe24feeb62e4ad8c9230564 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/212c142f55d24c6a9fe5acec845268b4 : SUCCESS in 3m 19s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/a8544bfd12b1413d8ebcf6c99160c5fc : SUCCESS in 14m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ff847e87ca914d05b14c98dd2e49f019 : SUCCESS in 12m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c58d473d6c264613a10849ae2b4bc479 : SUCCESS in 16m 21s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cb60e2ce6524f96a6d897982a52beab : SUCCESS in 50m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/aa99642b991f4c1fa684c6288f2c6a9f : SUCCESS in 1h 55m 19s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c7a1c7b2e70c4fee8963f34d51ac1145 : SUCCESS in 28m 25s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb5dd705ee3c4dbd97d7dfcf0b3c0992 : SUCCESS in 23m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3c29f9bcbe044e0a3c1f0c613bf3918 : SUCCESS in 3m 42s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2e40857772354151bbd60c0fa351c2f9 : SUCCESS in 1h 42m 55s","accounts_in_message":[],"_revision_number":1},{"id":"e4c536fba3fadea212e722273286523d9fae6aed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 07:42:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b09baa73ac1f3293c91be665bac77b55bef8bbf9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:39:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/229691d405524f3899f53e3c606a3a09 : SUCCESS in 4m 17s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/279db9bf8eb641f39e70e14c0315066b : SUCCESS in 3m 38s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2ec8d80a5e9440fcb8fefc0965503633 : SUCCESS in 15m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdc8c19b9d4a442994491c88d44acba3 : SUCCESS in 13m 06s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4ddb93548aa43ef82f759fbec6d2a30 : SUCCESS in 46m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/97fe5d31d29d46919963eea303cd50a0 : SUCCESS in 1h 32m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/72158df59bce48bb9e8f7a62cc752a08 : SUCCESS in 28m 34s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a02810ad3fda4012ac9cfa2a9ccd6b82 : SUCCESS in 24m 20s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e9ffcbdaa3249079e0b82882330c12f : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/13135b9d5fe142f295223c44cccd4bfe : SUCCESS in 1h 50m 52s","accounts_in_message":[],"_revision_number":1},{"id":"60cefc2ed2fcbf78c4184ae181b44787cfc6d092","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:39:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5d4db590f189242e501494d7acbe366fea79afe3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-15 09:42:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f2724290c604cb880fed7d292197c5d : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"12737dcda47b3a7bcfa9d5414935df586495bb10","revisions":{"12737dcda47b3a7bcfa9d5414935df586495bb10":{"kind":"REWORK","_number":1,"created":"2020-10-08 14:47:18.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/35/756835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/35/756835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/756835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/756835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/35/756835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/35/756835/1"}}},"commit":{"parents":[{"commit":"68019383e91ca149606c66c3f7462ee7b664d005","subject":"Update .gitreview for stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/68019383e91ca149606c66c3f7462ee7b664d005"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:47:16.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-10-08 14:47:16.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/victoria","message":"Update TOX_CONSTRAINTS_FILE for stable/victoria\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/victoria branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: Id1566ae7f851707f2243b6da2eb8290563369e5e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12737dcda47b3a7bcfa9d5414935df586495bb10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/12737dcda47b3a7bcfa9d5414935df586495bb10"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
