)]}'
{"id":"openstack%2Fpuppet-swift~758773","triplet_id":"openstack%2Fpuppet-swift~master~I1aca90eb467aba974a488301662465024e56e23b","project":"openstack/puppet-swift","branch":"master","topic":"swift3","hashtags":[],"change_id":"I1aca90eb467aba974a488301662465024e56e23b","subject":"Remove the swift::proxy::swift3 class","status":"MERGED","created":"2020-10-19 09:42:46.000000000","updated":"2020-10-19 17:33:56.000000000","submitted":"2020-10-19 17:30:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"758773-1603128624924-8149b301","meta_rev_id":"83d41ec9d625ee653c9fa44c9082c073ec7fa8ce","_number":758773,"virtual_id_number":758773,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-10-19 17:30:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":2,"date":"2020-10-19 15:17:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2020-10-19 14:35:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"all":[{"value":1,"date":"2020-10-19 15:17:27.000000000","_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-19 14:35:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2020-10-19 15:17:27.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2020-10-19 17:30:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d90307b88e41991f48acde032ebe35276c6a9e5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-10-19 09:42:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9f2b23069dd10764ad6d7918ee3880df3ce3992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 12:26:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d3253831f884c2399dde0894159a867 : SUCCESS in 6m 01s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17e4c0f6fced4dc1a31cbe5b3c3bdb33 : SUCCESS in 5m 41s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/23749c9603754b59976bc5a446e7c323 : SUCCESS in 18m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e3d5099faad499091c1a662ad62b526 : SUCCESS in 15m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6199ff3cccbb4d76b56a17d9b983ec64 : SUCCESS in 19m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/03d9b24074f741289c09a240f1b2dc76 : FAILURE in 29m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/835fdb133cde487fa408b1f922159626 : SUCCESS in 1h 33m 58s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7132c3edaa8b4a6bac3be885f37f1196 : SUCCESS in 29m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90b4cd6f520b4581a927948eb999b7dd : SUCCESS in 25m 33s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b25e704953ea43d280084b08600b66a8 : SUCCESS in 1h 31m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20d75a817c97461985bf343f33ea28aa : SUCCESS in 6m 43s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/dc3df965833b49a6b3b4a53aa18f109e : SUCCESS in 1h 50m 45s","accounts_in_message":[],"_revision_number":1},{"id":"480ee77b8624fce11c200bf7634d5d5ab9a346c8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2020-10-19 14:35:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6aa001a91b8d3403b75de199807863bc7ecc094d","author":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"date":"2020-10-19 15:17:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae4e23799d7d0ae3643e13549b1c576f96377653","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 15:17:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d89c565dbcb2e20c345ccf6f96b43dccc98e0df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:30:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"84d87e209016ec7cee4753b6a6b464cd661db4ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:30:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9a71a59087c43d0a2af4c5aad383f05 : SUCCESS in 6m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64d2a1210d864c08a7a2d3c423230a8b : SUCCESS in 8m 02s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e8d4d867d38c4e3982a3007a896c12d8 : SUCCESS in 17m 12s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69536631356f40fc857b026cda376841 : SUCCESS in 18m 56s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/29ba97bc4f414dc4bf2ac6ea574b631a : FAILURE in 32m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/af84ce00044441f5bf6c1310906b50fb : SUCCESS in 1h 32m 31s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b20bb1b4c43641bca81954408f72f515 : SUCCESS in 34m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92351fc736034f59b196152ced3a8930 : SUCCESS in 37m 12s (non-voting)\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/735a29bf40a54ea7b99aa5693f0482df : SUCCESS in 1h 21m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3c11e5e3b8145c8878fceb67c8758a5 : SUCCESS in 8m 15s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e7d95644be9243108d8fab661e2f5214 : SUCCESS in 1h 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"2bccf11da9b4682094d072cd18cb0179963d4a34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-19 17:33:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fea0df7302e34bb0b159bf40b450e6ae : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4ec1b631faa00aed2a939dcf2c76217024537ff","revisions":{"d4ec1b631faa00aed2a939dcf2c76217024537ff":{"kind":"REWORK","_number":1,"created":"2020-10-19 09:42:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/758773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/73/758773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/758773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/758773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/758773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/73/758773/1"}}},"commit":{"parents":[{"commit":"009afc806c126417f2f62eacd1c23d32a1be5390","subject":"Bump version for the start of Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/009afc806c126417f2f62eacd1c23d32a1be5390"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-19 09:41:38.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-10-19 09:42:33.000000000","tz":540},"subject":"Remove the swift::proxy::swift3 class","message":"Remove the swift::proxy::swift3 class\n\nThis patch removes the swift::proxy::swift3 class, which no longer has\nvalid implementation since it was deprecated[1].\n\n[1] f24fd8c73edb3e6892c68de072861ce7237ad6d8\nChange-Id: I1aca90eb467aba974a488301662465024e56e23b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4ec1b631faa00aed2a939dcf2c76217024537ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d4ec1b631faa00aed2a939dcf2c76217024537ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
