)]}'
{"id":"openstack%2Fpuppet-swift~768214","triplet_id":"openstack%2Fpuppet-swift~master~I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31","project":"openstack/puppet-swift","branch":"master","topic":"s3api","hashtags":[],"change_id":"I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31","subject":"Add missing parameters for ACL feature of s3api middleware","status":"MERGED","created":"2020-12-22 14:52:41.000000000","updated":"2021-01-04 17:51:04.000000000","submitted":"2021-01-04 17:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"768214-s3api","meta_rev_id":"10965f155208f16473d763b23792adb62ed9262e","_number":768214,"virtual_id_number":768214,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-04 17:49:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-01-04 17:51:04.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-01-04 14:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-01-04 14:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-22 16:23:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-27 18:47:22.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-01-04 14:47:38.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"e996eb59fb193ab6a9c56b3cb1a6b7ccf2afe9f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-22 14:52:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03c17d108f10f3c7c8a7ed887f4f389d758db795","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-22 14:53:20.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2b8e10f1e5f71503157ff6fb7571ed5eff2f0bdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-22 16:23:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83259cc2295e42a5b46e3efca4b3171f : SUCCESS in 3m 21s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4cd014a883943e09a1be5d556cf2abc : SUCCESS in 2m 52s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/587356cc445746ee9eed89cac935cf19 : FAILURE in 14m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ec6425653c6d4794815a59a739a843c9 : FAILURE in 10m 40s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4d87216b56c458780577bd999982550 : FAILURE in 14m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ad4dfeba2444d4abc53a3fb6b6366ee : SUCCESS in 57m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5f95144c6a24460d8cc66215292d3b75 : SUCCESS in 1h 27m 23s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/da89434320c5411fa930803acd9ce225 : SUCCESS in 24m 33s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ee1068f88cf42c5ad8158bec3610e89 : SUCCESS in 24m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f2c7dda26f444ea5aecd7350c468b63c : FAILURE in 58m 20s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36d39ac970104a0c849ae31c1bb915a0 : SUCCESS in 3m 19s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":2},{"id":"50429b876c0d48c2a4750a5ad34674396fe83ae6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-22 23:08:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5799afdefe880b2661f3e9e626eda84d178e278a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 01:58:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f5de58a1c904d2c9fed135cba7589c8 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aadd195a3e354baaae07ce4aa9d6135f : SUCCESS in 3m 35s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/873c1f7513194c1f8fe767b9a7199c1e : SUCCESS in 15m 17s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e65088bb75e4d269f00d69d4d49ab74 : SUCCESS in 8m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c1edd333e654642828140cbc1f11094 : SUCCESS in 14m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/baab55a1c5574598bc2fb966c9c79f80 : SUCCESS in 1h 00m 15s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/c51745b1b45447f4a9a8e12ca3d440d0 : SUCCESS in 1h 25m 45s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/b4f187f2e42844a3bd0ead8176d9fe48 : SUCCESS in 30m 29s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ad47a6430fa480ead589515f3797cc8 : SUCCESS in 21m 44s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d40e42a66d2c40f0981661eba60a7228 : SUCCESS in 2h 48m 38s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/60b2b0a6122645efb0cbd6104811f13c : SUCCESS in 1h 01m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9a8f70835a54769b1c2779bfba5f152 : SUCCESS in 3m 14s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8f4388d368d9407c890095e4dfa017f8 : SUCCESS in 1h 05m 36s","accounts_in_message":[],"_revision_number":3},{"id":"45078ca6e7d3527d43d12ed49ce3b11344584468","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-23 07:58:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8e6313d8f2862f1f66a7afba9ab6ba8e4e1f3ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-23 10:32:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2208b86581c46b1912764d36c607b81 : SUCCESS in 3m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9741aea5d5224e1d8a03d5208a8bc0db : SUCCESS in 3m 07s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/5ffd5519ecee4716aa2411dbd22370ce : SUCCESS in 12m 05s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a994826760e04e2fb1154635ea3bddf6 : SUCCESS in 10m 21s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f6fc31cde314102bfdf42fe3c2f6ccf : SUCCESS in 12m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2f5536553e84e91ba7271ff90cebbd9 : SUCCESS in 56m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/43e4051a9da54c67a0a812b57200f67c : SUCCESS in 1h 03m 02s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/15daafcd24d047018a9ff9baecac5826 : SUCCESS in 23m 17s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4ee17e8dbf94d4e843b859d876b3c5c : SUCCESS in 19m 19s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/0a1dcb66aa1d4ab6aaf3b40e139373c6 : SUCCESS in 2h 32m 15s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/75845ef14ebb4d8ca6e6a0cbc5883ef3 : SUCCESS in 1h 28m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a6ec49ddebc4fad858cce9b329ce4ee : SUCCESS in 3m 22s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/be0e20636c40460fba60b964e8dcaa12 : SUCCESS in 1h 16m 49s","accounts_in_message":[],"_revision_number":4},{"id":"9fb7392e1c9bc6d26502d74e2249aaf65ed9d8ca","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2020-12-27 18:47:22.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"a4c8bef373d98fce2a71aa363272e041a5aaa739","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2020-12-29 03:53:34.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"00f9c470bc2ade6d2794ea72aa82c4c8b68bf89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-29 06:18:24.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b958f6b5a8d447099910ed39592c5660 : SUCCESS in 2m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19fd410061704c04ba370ed9c9a4aac5 : SUCCESS in 2m 39s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/d1bb6b1f64ee4fc6a71c6188e91d5e6d : SUCCESS in 13m 44s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/acddfe0eaee642e99441f75e97025935 : SUCCESS in 11m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f0cfcc8dcbd0401199406408b0b53e5e : SUCCESS in 13m 24s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c7aef4961ae40dd8e2da15c07b17129 : SUCCESS in 51m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5e18711ca7834116a916d045f3380165 : SUCCESS in 1h 25m 43s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/15d1e0177aa34ecaac32f5a950f5ef10 : SUCCESS in 24m 19s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4298dca7603d44be9cb0c2bd706cf219 : SUCCESS in 21m 46s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/3e93419bd1144e399f849e2b6b325a38 : SUCCESS in 2h 19m 31s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e66ee269bc7743b4a4b11d4ba581e7a2 : SUCCESS in 1h 02m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/985bbda7873c4677b8ce9c90a4986ab3 : SUCCESS in 2m 59s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fa01889daad84cdab4d04c900c3794d2 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":5},{"id":"4419556d5dd993aa15f818f19532b308b0051045","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-04 14:47:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"04f4286ee70416770ad1f5244d0a9dc346c5170a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 14:47:51.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b2dc0c023a5d91dc729369495478e3c58b301e2d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 17:49:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1ba195530f89476fa59f1496c7646ae3 : SUCCESS in 3m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fac539432694cbe90ef582bb72b8807 : SUCCESS in 4m 22s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/06e4fd8b9b3c49d6be954fbf23f30649 : SUCCESS in 18m 54s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2593968af047452688e50a8caf8d347f : SUCCESS in 15m 54s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce39b72c5ce04f0bb9f86fe1527e4659 : SUCCESS in 55m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e4757f3a6db04a3180c623beca3e7bd9 : SUCCESS in 1h 33m 06s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/8fb13cfe68074964ae898c69da8cb290 : SUCCESS in 30m 11s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae37092a444945f69e8cd8ff510f4be6 : SUCCESS in 21m 39s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/7a937f09f9364930964a8ed542a017c7 : SUCCESS in 2h 57m 17s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a5e24175a20c480faaebb89333fc9c6c : SUCCESS in 1h 06m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2720ccca77304bf3ae42e746e8e5b30e : SUCCESS in 3m 29s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/22c4e0de32f34ad9b81605fb2fadfa3f : SUCCESS in 1h 57m 50s","accounts_in_message":[],"_revision_number":5},{"id":"d8a7fa0db119cb9be063431e33e23ec12838f688","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 17:49:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ed7a7834f7e9a0d57f8f4b425e876b51bb653151","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-04 17:51:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eb0bfdf05c347ea9351ce125f289a51 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c006f0adb3128b3ac54568fe0322f59ae1386b1e","revisions":{"20b227ba4b5dbb30bd43786c434cae644392eb03":{"kind":"REWORK","_number":1,"created":"2020-12-22 14:52:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/768214/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/768214/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/1"}}},"commit":{"parents":[{"commit":"2a1ca1cd5e25c023a6b6a0b354a8230834398275","subject":"Add support for the location parameter of s3api middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a1ca1cd5e25c023a6b6a0b354a8230834398275"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"subject":"Add support for parameters for ACL feature of s3api middleware","message":"Add support for parameters for ACL feature of s3api middleware\n\nChange-Id: I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/20b227ba4b5dbb30bd43786c434cae644392eb03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/20b227ba4b5dbb30bd43786c434cae644392eb03"}]},"branch":"refs/heads/master"},"ab59603b00479e440c2bacf9fc9cd88f7b44368d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-12-22 14:53:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/768214/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/768214/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/2"}}},"commit":{"parents":[{"commit":"2a1ca1cd5e25c023a6b6a0b354a8230834398275","subject":"Add support for the location parameter of s3api middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a1ca1cd5e25c023a6b6a0b354a8230834398275"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:53:05.000000000","tz":540},"subject":"Add missing parameters for ACL feature of s3api middleware","message":"Add missing parameters for ACL feature of s3api middleware\n\nChange-Id: I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ab59603b00479e440c2bacf9fc9cd88f7b44368d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ab59603b00479e440c2bacf9fc9cd88f7b44368d"}]},"branch":"refs/heads/master"},"626fc2fc78389d17bab2bcc921694cb4e1d29801":{"kind":"REWORK","_number":3,"created":"2020-12-22 23:08:38.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/768214/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/768214/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/3"}}},"commit":{"parents":[{"commit":"2a1ca1cd5e25c023a6b6a0b354a8230834398275","subject":"Add support for the location parameter of s3api middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a1ca1cd5e25c023a6b6a0b354a8230834398275"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 23:08:18.000000000","tz":540},"subject":"Add missing parameters for ACL feature of s3api middleware","message":"Add missing parameters for ACL feature of s3api middleware\n\nChange-Id: I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/626fc2fc78389d17bab2bcc921694cb4e1d29801"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/626fc2fc78389d17bab2bcc921694cb4e1d29801"}]},"branch":"refs/heads/master"},"fbf49dedf18525367078a4005f5f3bbf02262767":{"kind":"REWORK","_number":4,"created":"2020-12-23 07:58:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/768214/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/768214/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/4"}}},"commit":{"parents":[{"commit":"2a1ca1cd5e25c023a6b6a0b354a8230834398275","subject":"Add support for the location parameter of s3api middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a1ca1cd5e25c023a6b6a0b354a8230834398275"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-23 07:58:01.000000000","tz":540},"subject":"Add missing parameters for ACL feature of s3api middleware","message":"Add missing parameters for ACL feature of s3api middleware\n\nChange-Id: I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbf49dedf18525367078a4005f5f3bbf02262767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbf49dedf18525367078a4005f5f3bbf02262767"}]},"branch":"refs/heads/master"},"c006f0adb3128b3ac54568fe0322f59ae1386b1e":{"kind":"NO_CHANGE","_number":5,"created":"2020-12-29 03:53:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/14/768214/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/14/768214/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/14/768214/5"}}},"commit":{"parents":[{"commit":"cf0fa5d67980c24d67a72912d54a44f640bbdd4f","subject":"Add support for the location parameter of s3api middleware","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf0fa5d67980c24d67a72912d54a44f640bbdd4f"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-22 14:51:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2020-12-29 03:53:26.000000000","tz":540},"subject":"Add missing parameters for ACL feature of s3api middleware","message":"Add missing parameters for ACL feature of s3api middleware\n\nChange-Id: I4e6643d3c48e5d5591d60d26abd2ea037e0a6b31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c006f0adb3128b3ac54568fe0322f59ae1386b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c006f0adb3128b3ac54568fe0322f59ae1386b1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
