)]}'
{"id":"openstack%2Fpuppet-swift~772506","triplet_id":"openstack%2Fpuppet-swift~master~I58cf2bc21a0bc4265114b9daf3919a1a38f63649","project":"openstack/puppet-swift","branch":"master","topic":"objectexpirer-pipeline","hashtags":[],"change_id":"I58cf2bc21a0bc4265114b9daf3919a1a38f63649","subject":"object-expirer: Update the default pipeline","status":"MERGED","created":"2021-01-26 12:10:03.000000000","updated":"2021-01-28 13:17:25.000000000","submitted":"2021-01-28 13:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772506-objectexpirer-pipeline","meta_rev_id":"f4929203a356b3a3b02fd78456fe8b543f66dfc9","_number":772506,"virtual_id_number":772506,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 13:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-01-28 10:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-01-27 21:36:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-01-28 10:22:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 17:09:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 21:36:13.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-28 10:22:47.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"ac27a44d5c441ab4342c8b0d4d5271f71ba97371","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-26 12:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc7988a5178cc396ea07a123429df8364ebe38b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-26 14:05:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f25e919c0a86298a5f64a6d208d3f24bbe3dc865","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:09:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07476722a76e4f90a46301dc64f4d9f6 : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/203e6998863741ea9872439b4b245f7a : SUCCESS in 3m 40s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/3c93140abc6742249c0418ee99c81874 : FAILURE in 12m 08s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ee80f0df9ae44d2ab79169469f3bddf2 : FAILURE in 13m 58s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9db29b20d1a44056bbecec89d3f3aeb0 : FAILURE in 12m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/efcff82f0c57407e886f818e4d988372 : SUCCESS in 55m 10s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/66ad6d238f6c4ec6a04571509df11c25 : TIMED_OUT in 2h 00m 50s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f9570c12519a4d8fa9c6f0fbaf5cdb7b : FAILURE in 21m 56s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9e4b9f36355e4419bf04381f81ed5fca : FAILURE in 21m 54s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/dac03ff875874d0f9b43ac358e6c152a : SUCCESS in 2h 54m 48s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b2b08e080b31481f855d65ef1d72c4f8 : SUCCESS in 1h 27m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8057cdf7a45a4a33b401467d75cc66cf : SUCCESS in 3m 55s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6b99efea6f58497fa2be14218f6e4587 : SUCCESS in 1h 27m 06s","accounts_in_message":[],"_revision_number":2},{"id":"ebc2d0bbaab82beb48d8188e3106bedcfa6389c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-27 02:12:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e7927d38b96879b55810350215b4fd7398dec514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 05:38:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e8a76d70f23a4be4b8b2d579c81b07c3 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f883f58084cb4124ace75170ab6f26b9 : SUCCESS in 3m 53s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/eebe9a6022d24107ac53bc16530ce99a : SUCCESS in 12m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af25f0538b05413a8475cdfdb774dce2 : SUCCESS in 11m 24s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9905a9ac87fb4a5d9f3e6493aadb000b : SUCCESS in 14m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e970e6667a4b44fdbe4ed3b3b0123b91 : SUCCESS in 51m 02s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/eb682bf7730948b6b79d6b9e27d702c6 : SUCCESS in 1h 31m 10s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/f92c33f04929444a8b395a1ac19b5245 : SUCCESS in 36m 10s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ce0868c4e21472686b2b787dfb039f4 : SUCCESS in 21m 35s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/049bb25f539946b6ac07919d7617e8b1 : SUCCESS in 3h 24m 21s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/9f9414d6631e45b592ae22bb86352f92 : SUCCESS in 1h 04m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66ed275e6e304b239fadb39a9ada1ee7 : SUCCESS in 3m 34s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9523c4fc4c434ee29d385c5ed9df7312 : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"8161042247c25bfa4a52521697dee5fe835f939b","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-27 21:36:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e8b05dfba47a5cc1395016d1c061375f96c1d5c9","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-01-28 10:22:47.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4a4278db1c04a7cdd1d91042d36da758a027a3da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:23:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f1830e016041a6317472fb9ee310c34596f9de6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:15:53.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74afed4841d540b48daac4512406b351 : SUCCESS in 3m 51s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa37ef38d04d436882dfb018d50a8445 : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/f4356746699c4d3f88519573619157f3 : SUCCESS in 11m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53f4d1d2ea3d4297a20a8df2f09b2811 : SUCCESS in 9m 45s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b4d788c746a94de9bc55983511446ee8 : SUCCESS in 55m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f0c13e8aa96345a58201ecefd5a462b3 : SUCCESS in 1h 39m 30s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/d923f2f270e442f3a41681f20be5debc : SUCCESS in 27m 01s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/077b6a194c494cf2af5a13fe091466d4 : SUCCESS in 22m 11s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/68af842c26a14fefa3f3ce374ef96f8a : SUCCESS in 2h 45m 10s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d76ebd15353b434d92b2077e11d63f02 : SUCCESS in 1h 29m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c48988da5ad41939714c666816f13a8 : SUCCESS in 3m 44s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/004e85acfd4a4a95825151d7ae6deb6d : SUCCESS in 1h 27m 58s","accounts_in_message":[],"_revision_number":3},{"id":"d5941a19488e366e45ced71419d936c087880204","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:15:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"f4929203a356b3a3b02fd78456fe8b543f66dfc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:17:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a9a1ce2d4a6d4507bab0a838e9cf56fc : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"35bde997627d20c53cd7f330df050948d8af4809","revisions":{"4978ae83057630a7fdf047239fcb2edf1e28bd09":{"kind":"REWORK","_number":1,"created":"2021-01-26 12:10:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/772506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/772506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/1"}}},"commit":{"parents":[{"commit":"a57d5fe28a47c52b9137f000d3e923aeee1465a3","subject":"Merge \"Add TLS options for swift memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a57d5fe28a47c52b9137f000d3e923aeee1465a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 12:08:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 12:08:48.000000000","tz":540},"subject":"object-expirer: Update the default pipeline","message":"object-expirer: Update the default pipeline\n\nThis change updates the default pipeline of object-expirer, so that\nwe follow what is described in the default object-expirer.conf\nmaintained in swift repo.\n\nChange-Id: I58cf2bc21a0bc4265114b9daf3919a1a38f63649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4978ae83057630a7fdf047239fcb2edf1e28bd09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4978ae83057630a7fdf047239fcb2edf1e28bd09"}]},"branch":"refs/heads/master"},"4bb6eced8febcdd0a1cd608ca78d169bc9cded4d":{"kind":"REWORK","_number":2,"created":"2021-01-26 14:05:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/772506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/772506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/2"}}},"commit":{"parents":[{"commit":"a57d5fe28a47c52b9137f000d3e923aeee1465a3","subject":"Merge \"Add TLS options for swift memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a57d5fe28a47c52b9137f000d3e923aeee1465a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 12:08:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 14:05:33.000000000","tz":540},"subject":"object-expirer: Update the default pipeline","message":"object-expirer: Update the default pipeline\n\nThis change updates the default pipeline of object-expirer, so that\nwe follow what is described in the default object-expirer.conf\nmaintained in swift repo.\n\nChange-Id: I58cf2bc21a0bc4265114b9daf3919a1a38f63649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4bb6eced8febcdd0a1cd608ca78d169bc9cded4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4bb6eced8febcdd0a1cd608ca78d169bc9cded4d"}]},"branch":"refs/heads/master"},"35bde997627d20c53cd7f330df050948d8af4809":{"kind":"REWORK","_number":3,"created":"2021-01-27 02:12:19.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/06/772506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/772506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/772506/3"}}},"commit":{"parents":[{"commit":"a57d5fe28a47c52b9137f000d3e923aeee1465a3","subject":"Merge \"Add TLS options for swift memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a57d5fe28a47c52b9137f000d3e923aeee1465a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 12:08:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-27 02:12:10.000000000","tz":540},"subject":"object-expirer: Update the default pipeline","message":"object-expirer: Update the default pipeline\n\nThis change updates the default pipeline of object-expirer, so that\nwe follow what is described in the default object-expirer.conf\nmaintained in swift repo.\n\nChange-Id: I58cf2bc21a0bc4265114b9daf3919a1a38f63649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/35bde997627d20c53cd7f330df050948d8af4809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/35bde997627d20c53cd7f330df050948d8af4809"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
