)]}'
{"id":"openstack%2Fpuppet-swift~772530","triplet_id":"openstack%2Fpuppet-swift~master~Icde495557fef0ba4862ba2e4e79463c254b17de6","project":"openstack/puppet-swift","branch":"master","topic":"swift3-leftover","hashtags":[],"change_id":"Icde495557fef0ba4862ba2e4e79463c254b17de6","subject":"Remove the leftover of swift3 support","status":"MERGED","created":"2021-01-26 14:07:31.000000000","updated":"2021-01-28 13:08:23.000000000","submitted":"2021-01-28 13:08:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772530-swift3-leftover","meta_rev_id":"67d55bf382372d88ee0f514c6732890e048c142f","_number":772530,"virtual_id_number":772530,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-28 13:08:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-01-28 10:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-01-27 21:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":1,"date":"2021-01-28 10:22:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-26 17:06:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-27 21:36:01.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-01-28 10:22:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"}],"messages":[{"id":"bf17df9a1f3875c847de419d568c0b741344ce7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-26 14:07:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86a9cd4f5a621b4fba9c3086d5185a71d21cc8f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-01-26 14:07:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3e0f4aa0862c8dd61f8ecea5d9ee2e7e79a3c30c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 17:06:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ce0a4bbd657449429a3a15d6417f4a8e : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/683ad12f67ec4abc8f0bb5da4dba83ba : SUCCESS in 4m 12s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/26a835de8ec4403b863cebe1a1c9a397 : SUCCESS in 14m 38s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/268bf9db54664270b129da2e88d052e1 : SUCCESS in 10m 54s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/60036f0224904eb9bc4496ad11e217b6 : SUCCESS in 11m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c6b6afd212b4914bfbcd02b1f8f3d27 : SUCCESS in 50m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7d83cb9b285f4cdc921618296a13133b : SUCCESS in 1h 46m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/abf7c798c0b846509060555541ee8192 : SUCCESS in 34m 30s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a163918b982343e28c0261499402519e : SUCCESS in 25m 07s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/a5fb56045fd84b89b91b0025ddc256ba : SUCCESS in 2h 49m 21s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/a0e280a794b54ed9a5c37b42d93f4792 : SUCCESS in 1h 11m 31s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d5cc4ac5ba744e19a383dc5245a60123 : SUCCESS in 1h 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"609cbcacfe4154491df24e875b97ed96ac9cd233","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-01-27 21:36:01.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"feb50903549162f57e99aca60f59ecb081778f1b","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-01-28 10:22:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74924e809f2249d49e5ea278c5db5ff5958af9fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 10:22:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fbc0fe34a7ba8521f706eaa70a24c276495e6cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:08:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/123ce74bac6842b7bb76fd67f4520fe3 : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2a2baeaf9e59455799de7c0fb78fa3f5 : SUCCESS in 3m 08s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/b1fd3ad95b894f40876c281603b56835 : SUCCESS in 9m 56s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74af17dc9d7f478f8e5204711b69c2bd : SUCCESS in 8m 52s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2eec577d3ec54bd2baf00777c3b0e1a9 : SUCCESS in 1h 01m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73886d35a7ac4d56a0787b2810a275a1 : SUCCESS in 1h 40m 11s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/891ac6931d9740f5aad84a76f338d067 : SUCCESS in 26m 15s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9a34a75cf88468aa2b58b64ce6fc8d4 : SUCCESS in 26m 18s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/946127938a8c4b59bfabcd5d0904c81b : SUCCESS in 2h 41m 23s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/4520b845d82b4351a65910a8ee2317b8 : SUCCESS in 1h 14m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bbf03f1f3d904461b0100524d81b2bf3 : SUCCESS in 1h 26m 39s","accounts_in_message":[],"_revision_number":2},{"id":"67d55bf382372d88ee0f514c6732890e048c142f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 13:08:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"023863af52e69faae0e0eca87a52cd47f8ae7377","revisions":{"3c028ccdf615d4728dbb25a8d5f8d3cbb3f02fbc":{"kind":"REWORK","_number":1,"created":"2021-01-26 14:07:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/772530/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/772530/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/1"}}},"commit":{"parents":[{"commit":"a57d5fe28a47c52b9137f000d3e923aeee1465a3","subject":"Merge \"Add TLS options for swift memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a57d5fe28a47c52b9137f000d3e923aeee1465a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 14:06:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 14:06:14.000000000","tz":540},"subject":"Remove the leftover of swift3 support","message":"Remove the leftover of swift3 support\n\nThe support for swift3 has already been deprecated and removed[1], so\nwe can now remove the logic to ensure it\u0027s absent.\n\n[1] d4ec1b631faa00aed2a939dcf2c76217024537ff\n\nChange-Id: Icde495557fef0ba4862ba2e4e79463c254b17de6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3c028ccdf615d4728dbb25a8d5f8d3cbb3f02fbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3c028ccdf615d4728dbb25a8d5f8d3cbb3f02fbc"}]},"branch":"refs/heads/master"},"023863af52e69faae0e0eca87a52cd47f8ae7377":{"kind":"REWORK","_number":2,"created":"2021-01-26 14:07:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/30/772530/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/30/772530/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/30/772530/2"}}},"commit":{"parents":[{"commit":"a57d5fe28a47c52b9137f000d3e923aeee1465a3","subject":"Merge \"Add TLS options for swift memcached client\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a57d5fe28a47c52b9137f000d3e923aeee1465a3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 14:06:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-01-26 14:07:51.000000000","tz":540},"subject":"Remove the leftover of swift3 support","message":"Remove the leftover of swift3 support\n\nThe support for swift3 has already been deprecated and removed[1], so\nwe can now remove the logic to ensure it\u0027s absent.\n\n[1] d4ec1b631faa00aed2a939dcf2c76217024537ff\n\nChange-Id: Icde495557fef0ba4862ba2e4e79463c254b17de6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/023863af52e69faae0e0eca87a52cd47f8ae7377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/023863af52e69faae0e0eca87a52cd47f8ae7377"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
