)]}'
{"id":"openstack%2Fpuppet-swift~775606","triplet_id":"openstack%2Fpuppet-swift~stable%2Fvictoria~I4371d49df847236be18f1a33caa8b3f186f5eaf1","project":"openstack/puppet-swift","branch":"stable/victoria","topic":"memcached-cert-stable/victoria","hashtags":[],"change_id":"I4371d49df847236be18f1a33caa8b3f186f5eaf1","subject":"Add TLS options for swift memcached client","status":"MERGED","created":"2021-02-15 12:21:48.000000000","updated":"2021-02-23 20:59:16.000000000","submitted":"2021-02-23 20:57:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775606-memcached-cert-stable/victoria","meta_rev_id":"bb55ded80c7fa33790fb58b1d83bc296a7911e44","_number":775606,"virtual_id_number":775606,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-23 20:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-02-21 06:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-02-23 17:58:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-02-23 17:58:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"}],"reviewers":{"REVIEWER":[{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 15:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 16:29:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2021-02-19 16:29:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2021-02-19 16:29:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-02-19 16:29:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"2f8f139222b0992d06ff2297a17090af01373af8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-15 12:21:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40eb0eadf65275de8fcf2d084605c9205872b0d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/410af527abd142238c61bfb2dd4f2107 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa2d101d54624c6e8f39cd5195573126 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/443f98e93bc64cc4831c94ee1267d9e0 : SUCCESS in 12m 00s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/faa596cc86f1475ea9b25896ac621d9b : SUCCESS in 9m 55s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b99eaa097498464cb8919d83553e6cd0 : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eb5f2bf7f08d4c29897be98565ad61ea : FAILURE in 38m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e52cde8ecd6647e08bff6925dd186863 : SUCCESS in 1h 40m 25s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4323cb13d3cc4ba197a33a6668074e17 : SUCCESS in 26m 08s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bd932c8e92204ea7aca3c06afc999c53 : SUCCESS in 22m 09s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/42271e1a6afa433ebd542f87ed971065 : SUCCESS in 4m 07s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d1b440e581f54c11a1d647c22b72427b : SUCCESS in 2h 28m 43s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/00d941c94b93456a9232972fffb0cbd6 : SUCCESS in 1h 17m 51s","accounts_in_message":[],"_revision_number":1},{"id":"ef8228981b8c136a68f0c0807170aa5f314d63d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-15 16:23:37.000000000","message":"Topic set to memcached-cert-stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"091793b2bce03d90b068a55ccfce131e348e5c1c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-02-21 06:25:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2980e2f37da60eca16fe6383ad73b16e4e59e71","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 17:58:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c766d130a10449c705002a6875467805ac0794b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 17:58:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"120034cdf86c269f474ae484152be22390335ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:57:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a198566c01624d37a5f90c2684f71330 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61e885d96db34ce7a6bc593c08a1d890 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/939eea92a6914fdea669ed1dd0c7f186 : SUCCESS in 13m 31s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/433ec324dc654197bded33c80af65d0d : SUCCESS in 11m 49s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5573f5c0a2743838251e68027d6a3ba : FAILURE in 40m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/64a3de50f636439d96944b0653b59fd6 : SUCCESS in 1h 32m 12s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/1cf98dcbb8e74849af42a5af986b420a : SUCCESS in 22m 47s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6d8ddc703cb54441b3ed06b23336ceb1 : SUCCESS in 26m 12s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/303ea2ff5cd941c99b92e1d384b8e75f : SUCCESS in 4m 32s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/03653acd79d144d88db16978b8925fb3 : SUCCESS in 2h 56m 33s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a1fd44b6d38a4bd1a49f8317d9e797d1 : SUCCESS in 1h 47m 44s","accounts_in_message":[],"_revision_number":1},{"id":"bd80dbcee35c618c9e2b4c74e30402f55bc41543","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:57:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bb55ded80c7fa33790fb58b1d83bc296a7911e44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 20:59:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69fd00afeeae45a4ad1247b60f634276 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fbc7e6211e274b8b0f46b597cc2c365c6654c989","revisions":{"fbc7e6211e274b8b0f46b597cc2c365c6654c989":{"kind":"REWORK","_number":1,"created":"2021-02-15 12:21:48.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/06/775606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/06/775606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/775606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/775606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/06/775606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/06/775606/1"}}},"commit":{"parents":[{"commit":"2a0f3db4be7412b3cdf22e63a44088b2367fd819","subject":"Add content provider jobs for tripleo based jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2a0f3db4be7412b3cdf22e63a44088b2367fd819"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-08 13:05:06.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-02-15 12:21:32.000000000","tz":60},"subject":"Add TLS options for swift memcached client","message":"Add TLS options for swift memcached client\n\nThis patch specifies a set of options required to build a TLS context\nin the swift memcached client.\n(see https://review.opendev.org/c/openstack/swift/+/765552)\n\nChange-Id: I4371d49df847236be18f1a33caa8b3f186f5eaf1\n(cherry picked from commit 0688c8a576dd71af5a13b1da636de881e9be2224)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbc7e6211e274b8b0f46b597cc2c365c6654c989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fbc7e6211e274b8b0f46b597cc2c365c6654c989"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
