)]}'
{"id":"openstack%2Fpuppet-swift~775609","triplet_id":"openstack%2Fpuppet-swift~stable%2Ftrain~I4371d49df847236be18f1a33caa8b3f186f5eaf1","project":"openstack/puppet-swift","branch":"stable/train","topic":"memcached-cert-stable/train","hashtags":[],"change_id":"I4371d49df847236be18f1a33caa8b3f186f5eaf1","subject":"Add TLS options for swift memcached client","status":"MERGED","created":"2021-02-15 12:30:24.000000000","updated":"2021-04-13 13:15:29.000000000","submitted":"2021-04-13 13:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775609-memcached-cert-stable/train","cherry_pick_of_change":769918,"cherry_pick_of_patch_set":4,"meta_rev_id":"e672b8be9bdad803fbe37ca829fd2f44f8af06b2","_number":775609,"virtual_id_number":775609,"owner":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 13:13:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"all":[{"value":1,"date":"2021-04-13 10:33:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":2,"date":"2021-04-06 09:44:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-04-05 21:30:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},{"value":1,"date":"2021-04-06 09:44:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 14:15:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 19:25:37.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-03-21 11:54:19.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-03-24 13:52:12.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"CC"},{"updated":"2021-04-06 09:44:02.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-13 10:33:45.000000000","updated_by":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"reviewer":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"state":"REVIEWER"}],"messages":[{"id":"3eb2b0aff2d15a177a6d30828ffeeaf490550a8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-15 12:30:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"881b2afbfc98fd6a4ec80fbde8ca395727965081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 14:15:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7c58358c0a224ce6822ef11bad3f57f3 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/056ab6e046d24944b6e966c512ade64f : FAILURE in 5m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a68adf2094b64a068a205e97b7135ebd : FAILURE in 5m 35s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/d56950462999495fb3dc2e4cbb75e141 : FAILURE in 8m 14s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cdf96a6b45a44a3f9c987496414c8d6e : FAILURE in 11m 46s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/f9db8bf29e894eeaad4ea8823febe297 : FAILURE in 11m 40s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a14361f757ab4c90b7255e94568c378c : FAILURE in 12m 31s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cc197a93d9e467fb15b05bdf41bd757 : FAILURE in 23m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bf8952cc8df34c48a2c18738531450d7 : FAILURE in 5m 53s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/80a4f4d41dd14cafb67b1afa5c110e57 : FAILURE in 9m 48s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e869bc804e4c4a929eadbfe9bb3faf2f : FAILURE in 8m 35s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/62bb4630efaf41a093ff3e3a98dc4501 : FAILURE in 10m 30s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf10b4b912bf42558621eda727a377bd : FAILURE in 6m 40s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/9f531f3bb65548afb04d033dc359f164 : SUCCESS in 1h 35m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b6560634f5d4bffbf7f5f4ceff82382 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a8ec747a1a1079fc0ad390ab60b263b23e1293eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-15 16:25:46.000000000","message":"Topic set to memcached-cert-stable/train","accounts_in_message":[],"_revision_number":1},{"id":"9f5de9f0833dd2d30bae4f114fa4d8a38fce8550","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-18 14:44:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0aaf564f6a90a58ac75d9a42d9c7ae5f446db6d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:10:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9efe25921b0c430882590edb68559ac9 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5d552fa8756a4ffc9ea216a728383ffe : SUCCESS in 4m 28s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f5a750224e564591bc3f3ce6be0dcb0b : SUCCESS in 4m 55s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/87f8819815fe4f51b67fc444ff31dcdc : FAILURE in 8m 07s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/64bea8ac97be4708909bfbbab9c8e589 : FAILURE in 12m 04s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/ef7289b6a714481aa9f2d137f7168926 : FAILURE in 12m 04s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/214220c941de450cb8247cdbb15abf1e : FAILURE in 10m 06s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aac0ca289c9c4de2b7dfc190fdab66e6 : FAILURE in 20m 35s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ab5b7916b3e463b85344cc806bcb342 : FAILURE in 29m 06s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b023804f14104449a3185955eceaf7ac : SUCCESS in 1h 06m 23s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/253c44b2acd04a47a679ba02178c103e : SUCCESS in 1h 11m 48s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2cb5e27819954d618d363acba3103342 : SUCCESS in 27m 22s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d71a6550dda4661a12082c2b6756484 : FAILURE in 5m 58s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/7e61db0b58d1406c9a4e49c06a576052 : SUCCESS in 1h 18m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abc5cf1cd6e94f1ca044db688b7a55a8 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"2655479adbeb0d4e15df36ee459d6a833f29d8da","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-02-19 16:36:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"58ba8b9b3e58d97a344473b65271689160e115af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 18:10:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ee5cbf42bee484fabf8ade2f2be29ce : POST_FAILURE in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8889577cfef84efeb72396099d9d4f2a : POST_FAILURE in 5m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/637ad564b6064a2aaa6c9fdda087a2e1 : POST_FAILURE in 4m 41s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4483a6097760464c90bddfd2ed0f6e3b : POST_FAILURE in 22m 02s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/01e706cbf5df40c8a778ed8de2c20a2c : POST_FAILURE in 15m 56s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/17adcc5bdb3e44c28c6652b5fcf0e90c : POST_FAILURE in 15m 31s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9989da1bcbee4fb69e56d0a80ace3286 : POST_FAILURE in 9m 56s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c2510c2bbc94bb1bb41c628f902d91d : POST_FAILURE in 20m 07s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b95b9a5406e428badf4db3920b69ad7 : POST_FAILURE in 30m 21s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/8f456d27314c4fb8a4956b1e518b9b15 : POST_FAILURE in 52m 34s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/3dacbb8a2140452cb3cb403a1d56a512 : POST_FAILURE in 51m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/80c632dcd2b040f49234c465f84ccb81 : POST_FAILURE in 24m 57s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6539c7f440b34e19902d035fd1b2ec88 : POST_FAILURE in 7m 14s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/edf2f5e722f04c759ab784b4abd5c668 : POST_FAILURE in 1h 20m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6baa52b77e244db095126c4a6332ac02 : POST_FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"5bee9fcc55691bf51ef2454ae29ac62f53079014","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-02-23 19:25:37.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2423c453f28b8277a158318a15903cdc42c1ff18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 21:05:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1286b3da54bf4080a9b8d16807ef373d : SUCCESS in 4m 56s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f34765caed1d40208cba0dd60172d0e8 : SUCCESS in 5m 35s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2061adcc05bd4ab09fa16d5f6439f40f : SUCCESS in 5m 05s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/3a65ec87e5e9493f811a2afdc989b2b1 : FAILURE in 18m 59s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/e836a885e64d42c4aee9018c87c7a3e5 : FAILURE in 12m 40s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/ab7313f65e8a4ccead32916f51fd7a2a : FAILURE in 12m 17s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/afbb998aa74a45bcba842db0b87586fd : FAILURE in 10m 14s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ad413b7c0d24501a52945c824fa3629 : FAILURE in 20m 04s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/89c9a8f3970a4c93b18abfba0a4a000f : FAILURE in 30m 38s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/15423b5175d145d08c37a8a297507451 : SUCCESS in 59m 32s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1cb9abaf65704f90b17cb52b0ee25419 : SUCCESS in 53m 34s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d19619e7d0b54a0b87168c291091367b : SUCCESS in 28m 07s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cedd24a0a4cd4e97b4f1f5c924f2ec24 : FAILURE in 10m 05s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/d5f098a94c264d38a00960cd997ffce3 : SUCCESS in 1h 19m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b286ab913834dc784d7eb952d0a5acb : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":2},{"id":"9e55b633f915aa613fdd09e0994a16050826e15b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-03-01 20:53:06.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"45d665775815c98ae07effcac7c4d9484c2a90b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 22:10:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/30dfc985dcdb4dde801ab6056a69f8f6 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99c18f9a401345809bc1a01478ce175a : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d09f6090445144af91ef2ea16db9569b : SUCCESS in 5m 17s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/b8464fc08cf546b7befed370b4679d3c : FAILURE in 15m 47s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/589ea52d8cef493680acb82fd5d14708 : FAILURE in 15m 03s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/7bb76dfcc877476e8711897e5f9c15ac : FAILURE in 14m 09s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6e0e0d87a6f488298eee0ed90417efc : FAILURE in 9m 20s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af984c153dba473e960773c3b676754c : FAILURE in 24m 22s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/620f89f99b504c4a8a29455ecc8f8e28 : FAILURE in 27m 59s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c9b7957560ef44fab31d917c948f1290 : SUCCESS in 1h 00m 12s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/53f512bbe7a34572b22b55cb82b8215b : SUCCESS in 52m 00s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/2f7b001abef34c4481939be04582896a : SUCCESS in 27m 13s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d35b91dccbd04c7d8f4dc4acfbdce7bd : FAILURE in 6m 42s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/43c79b66523e43e0bbcfbbd8a08eeeef : SUCCESS in 1h 14m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7714624a0b0345fcb8f3dfe3671ff042 : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8171164b4b4f1fb201f4bc5de4dfd5b0b44b8d8a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-03-15 08:44:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"01a9ef6307b2dce088991c2479d0811358dce7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 10:09:11.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fcd76b4781c34b2bb464a27de3a71b3f : SUCCESS in 6m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bedf957aab1a436cacd119d47f4f52bf : SUCCESS in 5m 47s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbf88be38d58489eb878beef4f683895 : SUCCESS in 6m 21s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/4786d12f422345069997488fac3ed310 : FAILURE in 6m 56s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/b8ad0adf693b4971a4837a1ff8631f0c : FAILURE in 12m 56s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/aa4e6d8875e84c60872406b40d8072d5 : FAILURE in 13m 46s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/095fcfdf6c014beba662d5cdb2822856 : FAILURE in 9m 22s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7141a1e1c95a4e8d80e27a071f82aec2 : FAILURE in 17m 16s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9679cf82d8c943c2b707faef60dbbe7d : FAILURE in 31m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e5367af4f1074945b3131f474837e28f : SUCCESS in 1h 06m 20s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cfd95cd1b595433f8a5531f9b17887d9 : SUCCESS in 56m 31s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/f1017e4afbb44e379380f65886a39c1c : SUCCESS in 26m 05s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbfb6c26063647d59fd934dad84af859 : FAILURE in 7m 31s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/1a6dcb614abd4e18ae09b48b4422cf7d : SUCCESS in 1h 18m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2968b13788ec473691d1c93eb2491762 : SUCCESS in 4m 51s","accounts_in_message":[],"_revision_number":2},{"id":"fc9424b291bd58f2d08e13079d215743797cbe52","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-21 11:54:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d5e6df06c6eb2cf062c83d1d98c78404836f2fed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-21 12:47:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbcb509453e14f4f85d7cc2bcb2cbeec : SUCCESS in 4m 54s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/364260d8e7a94dbc929b54d84b4a8a5e : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e04b103e2524a3a80a655c1632bf7b6 : SUCCESS in 4m 08s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/0e3041c1ee574f9ca82dc4e67f665790 : FAILURE in 17m 42s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/cab016b7a7e741728f3d1d4783250ea2 : FAILURE in 9m 53s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/b3964b16381c46e6ab98bcd82a00470f : FAILURE in 11m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6badc9c12b2343439b40f835fc063b87 : FAILURE in 9m 16s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f3f465ab56484897bf54430ed3c53d15 : FAILURE in 13m 48s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7a1a833dacf4899b45253bb2db0b68b : FAILURE in 24m 31s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/b2cce342cb8042b3aa89dc589f11f68f : SUCCESS in 48m 19s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/abde066f1c31401da93a02ab281b3f38 : SUCCESS in 40m 47s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/bf55d8f883824204a2f3737cc49bcdd4 : SUCCESS in 23m 06s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e0f12eee0ec14633829a9fecd46a0240 : FAILURE in 6m 31s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/4ce30e5e08e5461ca4f2c41a27dba814 : SUCCESS in 51m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d769a9e8a9af4fa4afe671baff4175e6 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":2},{"id":"b80b2eb5c762b76e078d96ba41852662ea5c097d","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-24 13:52:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"fa50abd9d975368c795ae89fffcf8b80e9915b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 15:51:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/588d64c8000a4f76aa313628ac0ea2a0 : SUCCESS in 7m 16s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3324141c7d1748b6ab8cdb0436b3cd17 : SUCCESS in 5m 34s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44dae2848e14471fbdbb552e1fa16bb1 : SUCCESS in 5m 39s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/99cd72689079441b9166e8c892377de0 : FAILURE in 29m 23s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/5a467d33260c4c6994c69273fc438869 : FAILURE in 17m 04s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/02a0b37a0fe7467981b48b18e139a4ec : FAILURE in 17m 03s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ff12cbeee654bacab0cf9eb181fdcd8 : FAILURE in 11m 39s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b052bcfd0a64dcda3f6f95964a883c1 : POST_FAILURE in 20m 50s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5134f3a051dd4c66ba186aca7c022909 : FAILURE in 30m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/314aff952421436cbd040a97cd3b37b1 : SUCCESS in 1h 05m 13s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ab8b32420025463e9b0689eab667b0ae : SUCCESS in 59m 23s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/c61f3f2f9e7f4bfcbcc9ea7c7559ed11 : SUCCESS in 26m 46s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e81cbcb0f2504784afd8da0ae05baaaa : FAILURE in 8m 30s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/5347ad6a0c5341a59162d9117a57f3ee : SUCCESS in 1h 41m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfac4a06d5a1443f888a87a33a226a40 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":2},{"id":"65b1b442386ac881e67cf329c5eac5b96477cea5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-27 14:27:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"89fa7cf2bc97e29d130c9a2105c3b3c0ffb0af1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 15:50:38.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e548d9b1f8cf42ae94a771c436ed00ee : SUCCESS in 4m 59s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c587c3a42d994af5ac5cfd6145f1f8ea : SUCCESS in 4m 33s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f951f6580f194c98b9c629da42a4d274 : SUCCESS in 3m 59s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/ba302de95d4343bca7bb0b3e9c82164f : FAILURE in 19m 38s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/4e0193aefa4748739a90ee8333ca62e6 : FAILURE in 11m 49s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/6139b2aa9c38489a9831886e7b83b59b : FAILURE in 10m 59s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c633a6e80f849b28b14f6bdf5d80b81 : FAILURE in 8m 00s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a54e1e1f67cd400b874565c808c6242d : FAILURE in 16m 28s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9ac9239ef4e49fb8b52f6db98c98877 : FAILURE in 32m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/4ded4c3ef4294ac7888a29b1a2ebaae2 : SUCCESS in 1h 07m 17s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1a6705cfba9f45678dee89f7f32265f9 : SUCCESS in 40m 50s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/d75d64f1ff9d4be08e2557be75aa2025 : SUCCESS in 23m 38s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92764c1da4f344c0b8da877d9414e721 : FAILURE in 5m 43s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/e314c1f258d3437a8d3aeef6f0952bad : SUCCESS in 1h 22m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25de745cbc5841c5982b9cafd289e867 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d1ad500ee32042f78af788bd0109cb3b1647a8d5","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-03-31 08:49:47.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"76eb107b39f4a786de5ce2b52ba6d740217fabd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 10:15:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f2bcc83cb2e46c18687ef2c1aef0e09 : SUCCESS in 5m 34s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac49658486a24b6f91977558ec3e3c41 : SUCCESS in 5m 03s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e41d0c727a984e28a7f38129c38a373e : SUCCESS in 5m 40s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/dbc7eef2386246a4a99ea59705931e95 : FAILURE in 25m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/7ff9419102ed44e29ea615943f68fd5e : FAILURE in 11m 51s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/4d0090f8dd5c400f975dbb76410c766e : FAILURE in 13m 01s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0554ae68d6434fe9b8b641d6e661ddd4 : FAILURE in 10m 25s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/542763b6e12b4e6fb9e5c822a4199719 : FAILURE in 18m 27s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72205b960517429aaed3f15d889daac6 : FAILURE in 28m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/725e1674725d42168fffba49f15d0885 : SUCCESS in 1h 03m 37s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4dd9b2289423413599603b2210fb942d : SUCCESS in 1h 11m 34s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a964a80ce4854286835fd1f74bd496e0 : SUCCESS in 29m 37s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9e5ab33bdcb4e7ab0d950b602890ea4 : FAILURE in 9m 07s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/d92ea26bc28245f69161207bb15b9e30 : SUCCESS in 1h 13m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1472ea7c16184d9f9b28eab2e436eb23 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"bfd88f85d512410ce407ee49530dbdd74dfd3396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-31 13:31:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"084f4d2e13ae40f28f2c97d4adba89b417793acf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 15:14:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecea1e7958a14435b41445c67a64f628 : SUCCESS in 5m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b6ed47500534b89ac713afdd29355e0 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a625ef6b0ab4ddfbc06a905d658434f : SUCCESS in 4m 56s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/1c77cb9754a5469da237b73b5c1711b2 : SUCCESS in 20m 30s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/48f6aa35518747ea887ac3cc4d923c76 : SUCCESS in 14m 23s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/8f7854aeeee84366b7c90e10df2e2ace : SUCCESS in 14m 39s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c7872c8a244dd9a4657475f679ec90 : SUCCESS in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/eaf94ce012e548afb667fc2379d98746 : SUCCESS in 24m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/38a88475fbef4956a5d4e8cc485bbe04 : FAILURE in 31m 58s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/0470e5a3d3594bc3989510456bdd95a9 : SUCCESS in 54m 03s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/590e3c10e44b4ef6af922429ce6f8b0f : SUCCESS in 59m 15s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/1edb99b66a0846ca8255d31f01131164 : SUCCESS in 24m 10s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6262fbb822d2445d99c3c2f7d2a7f106 : FAILURE in 9m 12s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/f747293a18b543a486f111fc5f58deb6 : SUCCESS in 1h 25m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7d1736b5d9647a38d62ce5b700b4c93 : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":3},{"id":"8b802b21ddad83ecd22668f7e97086bdf5a33221","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-04-05 21:30:09.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9bdc5b78343bad720a6e59db728cce9693875832","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-06 09:44:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c88b682b4ab54fdbb1cc8f9067c36347cb42b7d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 09:44:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6e1f7c43d0ac5e9de5f62d8b22dff376596b8ed3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 11:06:01.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/475629a003b04a72a6112a5220f75ee6 : SUCCESS in 6m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/be0f89473328400284e60e1f91fd1408 : SUCCESS in 4m 57s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52434c8ccc37455da1011ccdae086672 : SUCCESS in 5m 37s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/cb8ff99371c34489be60ea73c3353925 : FAILURE in 9m 24s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6efdf051b804e95ab9b5d5067e1aae5 : SUCCESS in 13m 52s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19bec7bf07d3454f8cfad1a98e22cef6 : FAILURE in 26m 54s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/9870835dbe7843eea1fc3c9be926c58a : SUCCESS in 58m 48s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/38343c80701347b0b6c82782792ceffb : SUCCESS in 27m 15s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/41ad565496254a7691c645903f8af909 : SUCCESS in 1h 19m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95e6069c523f4084b28ae401fb5f9c08 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":3},{"id":"e7e2e0c1292f03ed37d41e9c6f48f66a04e82591","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-04-07 08:09:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c4eb1879819baf553d359dfca5e1d403a837f562","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:37:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cafdf70217344f1a5fa4cb1ce96e57c : SUCCESS in 5m 07s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5e4386abdea5444b921499adf3e018aa : SUCCESS in 6m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bc3f05ce2414f219f139fd2bcf2abac : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/2f70426a4dbd4ae58154b9da598eb814 : SUCCESS in 22m 40s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/43269e8632c84d20b019060e85b134fa : SUCCESS in 12m 28s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/a703bd45ee8849f9ab0f69c13713bf54 : SUCCESS in 9m 51s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ce6093444a245f8bf8211d5c53a5104 : SUCCESS in 9m 53s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/94a81ef393944838a8566645069927c0 : SUCCESS in 21m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4235c3b325de45de9a9560ebd3850bbb : FAILURE in 26m 10s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/c63774c049ef4d449708254e21db3017 : SUCCESS in 50m 08s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/079abf82e4f742819c50f6c4950ed5e9 : SUCCESS in 51m 21s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a2ffe8a8210e4dfe9613939ecdd0495e : SUCCESS in 22m 09s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6b0657e9ef3a491faf8a0d4a75720c7b : FAILURE in 6m 28s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/0a8a0650b3c945c28b28b4e10fc8488e : SUCCESS in 1h 22m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e2245f943e74302a9484db5bd3a6673 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":3},{"id":"bc6ffd7891491fee04840ddeaa8351a086edb12f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 09:38:04.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"98b58cfcc3fbd0f827c9afd6ae42e42cd8e64a8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 11:09:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e48c8c5498e246c29540a5a66d4636c1 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e7430aa95c04778b28f92bcb9d69fb2 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d883e02dfc9245f2a763317ae480786b : SUCCESS in 5m 13s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/5721e88a5de94999b2fc932f5058d17a : SUCCESS in 23m 10s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d72339f150e64f92a1950b88a1cb8f8e : SUCCESS in 10m 54s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9a2672513d04dfbb5135fa5591133f6 : FAILURE in 6m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ca7def4cef01470fb15e351f249041ae : SUCCESS in 1h 02m 42s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/0be109480197461089b348a5add9a612 : SUCCESS in 27m 04s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/90441f6c1b854b50b2483251531f570c : SUCCESS in 1h 25m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28c7929b71d9475ab21b0b31dfae8032 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":3},{"id":"77a3b1a3fc5e1ff316397619d350247a59a61275","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-04-09 07:34:43.000000000","message":"Patch Set 3:\n\nI wonder why didn\u0027t this one merge 🤔","accounts_in_message":[],"_revision_number":3},{"id":"989daa928fc7c77c7b77b49bcf427732df9966fd","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2021-04-09 07:36:36.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e I wonder why didn\u0027t this one merge 🤔\n\nlooks like it did - gerrit didn\u0027t get updated","accounts_in_message":[],"_revision_number":3},{"id":"0c22f495b64476fa12692010360fe4e832b9d256","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-04-13 10:33:45.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"98b9bd3670dc31fcb1bae5c7401959025ba1fc9b","author":{"_account_id":27954,"name":"Moisés Guimarães de Medeiros","email":"guimaraes@pm.me","username":"moguimar"},"date":"2021-04-13 10:34:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"b608cf1f3efea8646c2a94317b86c075dc88213c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 12:01:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44bbebe4f22940dba6bdafd03eca9a73 : SUCCESS in 4m 33s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98e7ec80ecfc439abd18f586e7ce0a8b : SUCCESS in 5m 04s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/728c188548294d00a142bf83397f5c7a : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/72b7c4efbd324a8891bc56d2062eb30e : SUCCESS in 19m 04s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/97966f9043594e0bb1ee4ef7f936c5cf : SUCCESS in 12m 42s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/b4b049836e9747b9b586e573d05db915 : SUCCESS in 11m 52s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/613989ca2c3b459f8cb08d4ad0f5464e : SUCCESS in 9m 57s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58070913f4ce4b16ac54c2de0e6f3edf : SUCCESS in 17m 57s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a104891ec7e4e238a1755551698f54d : FAILURE in 32m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/e6d12d3ea5d84066a228d2faf6118f03 : SUCCESS in 51m 04s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/27b13744b67441748edf9a4fd827fcb9 : SUCCESS in 57m 40s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/6e27bb55f01e4eed8219aa32e07891ad : SUCCESS in 24m 17s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d51813f7221a4c469e37e48e86ea117e : FAILURE in 7m 14s (non-voting)\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/b2a6652b49264b1eadb5f2ab434707f3 : SUCCESS in 1h 25m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ed928aa95564435901f371dbc90ea16 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":3},{"id":"66c8cb85b329fdd27184696ce356bc2e7c5edae1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 12:02:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"41d9a51adec446f11a0eb25f42f0200e8c5ae012","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 13:13:15.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1167904395fe488c9ee7c87738e6c3b3 : SUCCESS in 5m 08s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23cd0ae1a313403492f12140cfdf85b3 : SUCCESS in 4m 40s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b82819410b84225977101402ecd3215 : SUCCESS in 4m 24s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/c7ed79b1ad384d14864c15b866872fe8 : SUCCESS in 17m 23s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/15e24a61cfb54a8b8fd40a4d65182fd6 : SUCCESS in 10m 28s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5106d076cca241a4bb07cded8c473d94 : FAILURE in 30m 56s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/ee7d5ddee54f48679bff11bccdc513df : SUCCESS in 50m 03s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/24fc31437079477ea7cbc18eea6bf0c2 : SUCCESS in 23m 35s\n- tripleo-ci-centos-7-content-provider https://zuul.opendev.org/t/openstack/build/0f2f0c343f254d3487075ba643f5d84b : SUCCESS in 1h 09m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66b67e29512049bebd5a27429d9cbaa7 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"e7cde43e96be1eabdbf03d9f1916af11691007df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 13:13:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"e672b8be9bdad803fbe37ca829fd2f44f8af06b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 13:15:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e640ad2b3e4a48a8bcc6ae0e61e64e45 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"feb9705870d6654f4db1e0b968db0d964fa36769","revisions":{"889e5f86e7957ce111c44012b9c5900e4deb5452":{"kind":"REWORK","_number":1,"created":"2021-02-15 12:30:24.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/09/775609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/09/775609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/1"}}},"commit":{"parents":[{"commit":"f37bc1a3024e14ff5a0b49b1ab7b09c14dc44751","subject":"Build containers for single consumer job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f37bc1a3024e14ff5a0b49b1ab7b09c14dc44751"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-08 13:05:06.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-02-15 12:30:12.000000000","tz":60},"subject":"Add TLS options for swift memcached client","message":"Add TLS options for swift memcached client\n\nThis patch specifies a set of options required to build a TLS context\nin the swift memcached client.\n(see https://review.opendev.org/c/openstack/swift/+/765552)\n\nChange-Id: I4371d49df847236be18f1a33caa8b3f186f5eaf1\n(cherry picked from commit 0688c8a576dd71af5a13b1da636de881e9be2224)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/889e5f86e7957ce111c44012b9c5900e4deb5452"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/889e5f86e7957ce111c44012b9c5900e4deb5452"}]},"branch":"refs/heads/stable/train"},"9d429b839d9ec3d4a8b2b9a3bfeeecb443fe151a":{"kind":"REWORK","_number":2,"created":"2021-02-18 14:44:32.000000000","uploader":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"ref":"refs/changes/09/775609/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/09/775609/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/2"}}},"commit":{"parents":[{"commit":"f37bc1a3024e14ff5a0b49b1ab7b09c14dc44751","subject":"Build containers for single consumer job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f37bc1a3024e14ff5a0b49b1ab7b09c14dc44751"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-08 13:05:06.000000000","tz":60},"committer":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-02-18 14:44:20.000000000","tz":60},"subject":"Add TLS options for swift memcached client","message":"Add TLS options for swift memcached client\n\nThis patch specifies a set of options required to build a TLS context\nin the swift memcached client.\n(see https://review.opendev.org/c/openstack/swift/+/765552)\n\nChange-Id: I4371d49df847236be18f1a33caa8b3f186f5eaf1\n(cherry picked from commit 0688c8a576dd71af5a13b1da636de881e9be2224)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9d429b839d9ec3d4a8b2b9a3bfeeecb443fe151a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9d429b839d9ec3d4a8b2b9a3bfeeecb443fe151a"}]},"branch":"refs/heads/stable/train"},"feb9705870d6654f4db1e0b968db0d964fa36769":{"kind":"REWORK","_number":3,"created":"2021-03-31 13:31:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/09/775609/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/09/775609/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/09/775609/3"}}},"commit":{"parents":[{"commit":"d36c510110983dc80bdf49c1aeb98bdf46fe3170","subject":"Add parameter memcache_max_connections","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d36c510110983dc80bdf49c1aeb98bdf46fe3170"}]}],"author":{"name":"Grzegorz Grasza","email":"xek@redhat.com","date":"2021-01-08 13:05:06.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-31 13:31:54.000000000","tz":0},"subject":"Add TLS options for swift memcached client","message":"Add TLS options for swift memcached client\n\nThis patch specifies a set of options required to build a TLS context\nin the swift memcached client.\n(see https://review.opendev.org/c/openstack/swift/+/765552)\n\nChange-Id: I4371d49df847236be18f1a33caa8b3f186f5eaf1\n(cherry picked from commit 0688c8a576dd71af5a13b1da636de881e9be2224)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/feb9705870d6654f4db1e0b968db0d964fa36769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/feb9705870d6654f4db1e0b968db0d964fa36769"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
