)]}'
{"id":"openstack%2Fpuppet-swift~775631","triplet_id":"openstack%2Fpuppet-swift~master~I2049b8078c5205a56fc2ffcb1f2cc6453d862cb9","project":"openstack/puppet-swift","branch":"master","topic":"fstab-fact","hashtags":[],"change_id":"I2049b8078c5205a56fc2ffcb1f2cc6453d862cb9","subject":"Check for commented fstab entries","status":"ABANDONED","created":"2021-02-15 13:54:59.000000000","updated":"2023-03-23 04:28:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"975db227cc400e35b2e0b0a6aada2399c4fbaf7a","_number":775631,"virtual_id_number":775631,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-15 17:25:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"all":[{"value":1,"date":"2021-02-15 15:51:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 15:51:15.000000000","updated_by":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"reviewer":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"state":"REVIEWER"},{"updated":"2021-02-15 17:25:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4e9cb148aee51b86b8e0c5ee6603f683f57e1f52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-02-15 13:54:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a31e2437c7f1bd6fb5824c3f6ef390514f13d21d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-02-15 14:10:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"96b66407aabcd96cf01c6188867db8062fa402e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-02-15 14:12:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c6c6948140ab38499df35ce9d0dc93f000a7df95","author":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"date":"2021-02-15 15:51:15.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"07fbd1a07bc0a7f312503d6be375af15941e3211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:25:52.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b8f1e36d124f4ac8a167aa62725e118b : SUCCESS in 3m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f22ff4f49ff440a18a56ff7ad4a09a4b : SUCCESS in 3m 15s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/2cbd7472e15f449f92640d6be72069e0 : FAILURE in 11m 59s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/dc5f7b29299e455c9ce2d1ad3e985c10 : FAILURE in 11m 59s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0867969fbb504bddaec62697a8a08a81 : FAILURE in 9m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6117c73abcd484cbe7f9b879a9fda7b : FAILURE in 12m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/25cd78ae66e4497e857f89f83c6a06b5 : FAILURE in 35m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/799e9dc831604155a6a4250f10a8c4fa : SUCCESS in 1h 42m 14s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c834a20b50bc427b8c8ebc7e49bdf9f9 : SUCCESS in 1h 37m 56s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/19b81dadf90c4d858734a0f7298bb2b2 : SUCCESS in 25m 54s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/33cd6a3fb5754cf581ecaf3ee4ecfcb5 : SUCCESS in 23m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/01be478f2d364d6494438c3bf67ab7ff : SUCCESS in 2h 37m 24s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/2766ad0955ea4348aa60f1f44adb2696 : SUCCESS in 1h 11m 41s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f4fb96e28ddc445ab085bf169de70b91 : SUCCESS in 1h 24m 47s","accounts_in_message":[],"_revision_number":3},{"id":"71ad6e5dab33982ffb989d935c52504f6977ccc3","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-02-17 10:57:23.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c428f3e9b737a38fd22bfeba22dae7a0f88f5887","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 13:59:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11fefeea45844ab2846eebf26a0ac05e : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/92d07f956ba24d8192d92dc58b387bbb : SUCCESS in 3m 32s\n- puppet-openstack-unit-6.14-centos-8 https://zuul.opendev.org/t/openstack/build/e590b09b587942c8a6d11dcb56574c0b : FAILURE in 12m 02s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8f13b3b28ee47499994266c080d7955 : FAILURE in 14m 53s (non-voting)\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/209e82ca3dd747d9a71fba713b95064f : FAILURE in 11m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/589e9d8bce494096a43e7484f1e0aaf8 : FAILURE in 14m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4d45afab41984ba99d1e945e87da69af : FAILURE in 37m 57s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f4908ad04a7e445cb0414c3661a8ba10 : SUCCESS in 1h 38m 12s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bac2caad06f2480face25bf529632bb5 : SUCCESS in 1h 30m 47s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/553c5eb6594c46ddb467f31df0faa9e3 : SUCCESS in 29m 52s\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/256af88d27784eba9ab8b28de11f8702 : SUCCESS in 22m 42s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c81bdf2ec88448b38f242277b1963f01 : SUCCESS in 2h 41m 22s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/cc1e3776788a4864b7fd87b16683621e : SUCCESS in 1h 09m 05s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6e878595df22474ba3abb61734095fba : SUCCESS in 1h 27m 44s","accounts_in_message":[],"_revision_number":3},{"id":"975db227cc400e35b2e0b0a6aada2399c4fbaf7a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-03-23 04:28:32.000000000","message":"Abandoned\n\nI\u0027ll abandon this because this failed in CI and has been inactive for long","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b2e2002fb70ed70844cca8596153c13a4b7e6d1a","revisions":{"42fddd596c53b924a217013e265364688d9f722b":{"kind":"REWORK","_number":1,"created":"2021-02-15 13:54:59.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/31/775631/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/31/775631/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/1"}}},"commit":{"parents":[{"commit":"3bcbf44125c5e96c66c55f6cf4dc5017524937f3","subject":"Merge \"object-expirer: Update the default pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3bcbf44125c5e96c66c55f6cf4dc5017524937f3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 13:52:17.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 13:54:55.000000000","tz":60},"subject":"Check for commented fstab entries","message":"Check for commented fstab entries\n\nUnder normal production operations, it\u0027s quite common to have a broken\nhard drive, unfortunately, puppet-swift in this case, keeps trying to\nformat and mount the drive. This patch prevents this, by looking for\ncommented out entries in /etc/fstab with a custom Ruby fact.\n\nChange-Id: I2049b8078c5205a56fc2ffcb1f2cc6453d862cb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/42fddd596c53b924a217013e265364688d9f722b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/42fddd596c53b924a217013e265364688d9f722b"}]},"branch":"refs/heads/master"},"d7a60cbc9c8d3f1488511ed6454a852bf6f73945":{"kind":"REWORK","_number":2,"created":"2021-02-15 14:10:31.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/31/775631/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/31/775631/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/2"}}},"commit":{"parents":[{"commit":"3bcbf44125c5e96c66c55f6cf4dc5017524937f3","subject":"Merge \"object-expirer: Update the default pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3bcbf44125c5e96c66c55f6cf4dc5017524937f3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 13:52:17.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 14:10:16.000000000","tz":60},"subject":"Check for commented fstab entries","message":"Check for commented fstab entries\n\nUnder normal production operations, it\u0027s quite common to have a broken\nhard drive, unfortunately, puppet-swift in this case, keeps trying to\nformat and mount the drive. This patch prevents this, by looking for\ncommented out entries in /etc/fstab with a custom Ruby fact.\n\nChange-Id: I2049b8078c5205a56fc2ffcb1f2cc6453d862cb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d7a60cbc9c8d3f1488511ed6454a852bf6f73945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d7a60cbc9c8d3f1488511ed6454a852bf6f73945"}]},"branch":"refs/heads/master"},"b2e2002fb70ed70844cca8596153c13a4b7e6d1a":{"kind":"REWORK","_number":3,"created":"2021-02-15 14:12:44.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/31/775631/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/31/775631/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/31/775631/3"}}},"commit":{"parents":[{"commit":"3bcbf44125c5e96c66c55f6cf4dc5017524937f3","subject":"Merge \"object-expirer: Update the default pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3bcbf44125c5e96c66c55f6cf4dc5017524937f3"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 13:52:17.000000000","tz":60},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2021-02-15 14:12:29.000000000","tz":60},"subject":"Check for commented fstab entries","message":"Check for commented fstab entries\n\nUnder normal production operations, it\u0027s quite common to have a broken\nhard drive, unfortunately, puppet-swift in this case, keeps trying to\nformat and mount the drive. This patch prevents this, by looking for\ncommented out entries in /etc/fstab with a custom Ruby fact.\n\nChange-Id: I2049b8078c5205a56fc2ffcb1f2cc6453d862cb9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b2e2002fb70ed70844cca8596153c13a4b7e6d1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/b2e2002fb70ed70844cca8596153c13a4b7e6d1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
