)]}'
{"id":"openstack%2Fpuppet-swift~782347","triplet_id":"openstack%2Fpuppet-swift~master~I7e3604aaa8c2254f6c13193c9e27b259371d81d3","project":"openstack/puppet-swift","branch":"master","topic":"config","hashtags":[],"change_id":"I7e3604aaa8c2254f6c13193c9e27b259371d81d3","subject":"WIP: Use *_config resources to manage config files","status":"ABANDONED","created":"2021-03-23 02:08:59.000000000","updated":"2022-05-22 23:58:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9ce31423f8fca34ff66325a84868ef4e3b873e96","_number":782347,"virtual_id_number":782347,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-23 03:40:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-23 03:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afd1b75fc9593aa73d9a8f3660e122b7f9efd05f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-23 02:08:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58b2ae52e31a0272eb68f52ae99c42573012860f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 03:40:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f84d0d72d8442f384ae34426a5160f1 : FAILURE in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d153681b8704cdd808c3e30262b2373 : FAILURE in 3m 26s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/21aad40bfe9a4756b08569f2f3c2ffe6 : SUCCESS in 13m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5b63da20d884811ad597bf38dcbd77d : SUCCESS in 12m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e935f54ad7bc4addb7563ae12de9805d : SUCCESS in 8m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9791cb3072b847638cab74d4368222b6 : SUCCESS in 54m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/11f75deba8b4449d96198c273f5ae0dc : FAILURE in 1h 26m 24s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/90729eb91339499b8ed1517044d41be8 : SUCCESS in 27m 32s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a5ccfb252dd24b249cf5e2d7f1c4655a : FAILURE in 22m 16s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cdd48b70c1654f7f883082ac1be36517 : FAILURE in 1h 04m 59s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/None : SKIPPED\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/None : SKIPPED","accounts_in_message":[],"_revision_number":1},{"id":"9ce31423f8fca34ff66325a84868ef4e3b873e96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-05-22 23:58:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ea5c0881d5ebff396ccbb4a611ff8ae8709a524a","revisions":{"ea5c0881d5ebff396ccbb4a611ff8ae8709a524a":{"kind":"REWORK","_number":1,"created":"2021-03-23 02:08:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/47/782347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/47/782347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/47/782347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/47/782347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/47/782347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/47/782347/1"}}},"commit":{"parents":[{"commit":"ede897b1125d382f2aa886be3e5a74eefdfbaeb1","subject":"Fix name of ceilometer middleware package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ede897b1125d382f2aa886be3e5a74eefdfbaeb1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-23 02:07:41.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-23 02:08:51.000000000","tz":540},"subject":"WIP: Use *_config resources to manage config files","message":"WIP: Use *_config resources to manage config files\n\nChange-Id: I7e3604aaa8c2254f6c13193c9e27b259371d81d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea5c0881d5ebff396ccbb4a611ff8ae8709a524a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ea5c0881d5ebff396ccbb4a611ff8ae8709a524a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
