)]}'
{"id":"openstack%2Fpuppet-swift~782676","triplet_id":"openstack%2Fpuppet-swift~master~I2ed6b5be63923b01e300e8bc07e716c60ab4751a","project":"openstack/puppet-swift","branch":"master","hashtags":[],"change_id":"I2ed6b5be63923b01e300e8bc07e716c60ab4751a","subject":"Add object-expirer section to object-server.conf","status":"ABANDONED","created":"2021-03-24 09:40:11.000000000","updated":"2021-03-24 15:53:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a20c4a6d7812fb4ba1baa0440168ef44fd602e4a","_number":782676,"virtual_id_number":782676,"owner":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-24 12:30:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2021-03-24 09:45:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 09:45:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-03-24 12:30:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f0491afc5da70be2ef5467e14602689e10ed03a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-24 09:40:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b293b8140a09a251c2233ca9f154f552544d7dcb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 09:45:11.000000000","message":"Patch Set 1: Code-Review-1\n\nIf all contents in current object-expirer.conf is migrated to object-server.conf then we should update swift::objectexpirer which is currently putting options to that old file...\n\nThat switch might be able to be implemented easily by adidng usage of swift_object_config, but a problem might be that because we are using template to generate object-server.conf using template injecting parameters using swift_object_config might break Idempotency.\n\nI recently started working to replace current usage of templates by ini providers. Maybe we would need that one as well...\nhttps://review.opendev.org/c/openstack/puppet-swift/+/782347","accounts_in_message":[],"_revision_number":1},{"id":"670a8258f310e134b969e0c457036fb5b4ce664b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-24 11:02:33.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e If all contents in current object-expirer.conf is migrated to object-server.conf then we should update swift::objectexpirer which is currently putting options to that old file...\n\u003e \n\u003e That switch might be able to be implemented easily by adidng usage of swift_object_config, but a problem might be that because we are using template to generate object-server.conf using template injecting parameters using swift_object_config might break Idempotency.\n\u003e \n\u003e I recently started working to replace current usage of templates by ini providers. Maybe we would need that one as well...\n\u003e https://review.opendev.org/c/openstack/puppet-swift/+/782347\n\nThe problem with moving to a new config file is promotion and backwards compatibility. We need a way to maintain backwards compatibility until we get the new package promoted so need to have valid configs for both the new and the old config files.","accounts_in_message":[],"_revision_number":1},{"id":"3c27738de30c307bf13114df6f82b1e074d917ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-24 12:28:59.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \n\u003e \u003e If all contents in current object-expirer.conf is migrated to object-server.conf then we should update swift::objectexpirer which is currently putting options to that old file...\n\u003e \u003e \n\u003e \u003e That switch might be able to be implemented easily by adidng usage of swift_object_config, but a problem might be that because we are using template to generate object-server.conf using template injecting parameters using swift_object_config might break Idempotency.\n\u003e \u003e \n\u003e \u003e I recently started working to replace current usage of templates by ini providers. Maybe we would need that one as well...\n\u003e \u003e https://review.opendev.org/c/openstack/puppet-swift/+/782347\n\u003e \n\u003e The problem with moving to a new config file is promotion and backwards compatibility. We need a way to maintain backwards compatibility until we get the new package promoted so need to have valid configs for both the new and the old config files.\n\nIf I understand the change mentioned in the commit message correctly, backward incompatible change was made in RDO side because it removed  object-expirer.conf on which puppet-swift still depends.\n\nWe would be able make object-expirer run with adding that section to object-server.conf but then the process will be started without any configurations, thus I doubt how useful it is to add this now to just unblock RDO promotion.\n\nIMO we should not merge that change in RDO side until we add proper logic to put object-expirer options in object-server.conf . Note that it seems that Ubuntu Cloud Archives Victoria still uses object-expirer.conf . This means we would need to keep both configurations until we hear clear plans from each distro to switch to object-server.conf.\n\nhttps://zuul.opendev.org/t/openstack/build/048215afe33544209d4c5298e4b80b6a/log/logs/systemctl.txt#2489-2509","accounts_in_message":[],"_revision_number":1},{"id":"8b63120377dbcb4c47f1e89fd3dc114a1902b4f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:30:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e261edba33064cecaac10c9bf9c4dc08 : SUCCESS in 4m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a6b6a4d42a04bbeade98e6abdf84c0c : SUCCESS in 3m 51s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/4637e06eb8da40788e8f7ddb1f55a97c : SUCCESS in 12m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad2d1a418c55453495f41235a037cdd1 : SUCCESS in 12m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a9702c2d638454ab5f6c8d4be10bd5a : SUCCESS in 13m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bdf9152a21b244aa9845b6f93d00c569 : SUCCESS in 59m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/77b17102f7854adb8c6e133b878ca8e9 : SUCCESS in 1h 46m 41s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/e4de25108ebb4c669417631f52b837da : SUCCESS in 26m 45s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7944d89bb3f9424e848e524b0c2eedfd : FAILURE in 36m 38s (non-voting)\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/66e652b6171f45d1b8e8eab9691866e2 : SUCCESS in 2h 44m 52s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/bd2be6191bad44709bab5859f501f474 : SUCCESS in 1h 24m 08s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/578baa02b08a42f2b6485f9274de8b39 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"824e387e26ea468458ee777667fab5ac7155c641","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-24 12:58:34.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 1: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e If all contents in current object-expirer.conf is migrated to object-server.conf then we should update swift::objectexpirer which is currently putting options to that old file...\n\u003e \u003e \u003e \n\u003e \u003e \u003e That switch might be able to be implemented easily by adidng usage of swift_object_config, but a problem might be that because we are using template to generate object-server.conf using template injecting parameters using swift_object_config might break Idempotency.\n\u003e \u003e \u003e \n\u003e \u003e \u003e I recently started working to replace current usage of templates by ini providers. Maybe we would need that one as well...\n\u003e \u003e \u003e https://review.opendev.org/c/openstack/puppet-swift/+/782347\n\u003e \u003e \n\u003e \u003e The problem with moving to a new config file is promotion and backwards compatibility. We need a way to maintain backwards compatibility until we get the new package promoted so need to have valid configs for both the new and the old config files.\n\u003e \n\u003e If I understand the change mentioned in the commit message correctly, backward incompatible change was made in RDO side because it removed  object-expirer.conf on which puppet-swift still depends.\n\u003e \n\u003e We would be able make object-expirer run with adding that section to object-server.conf but then the process will be started without any configurations, thus I doubt how useful it is to add this now to just unblock RDO promotion.\n\u003e \n\u003e IMO we should not merge that change in RDO side until we add proper logic to put object-expirer options in object-server.conf . Note that it seems that Ubuntu Cloud Archives Victoria still uses object-expirer.conf . This means we would need to keep both configurations until we hear clear plans from each distro to switch to object-server.conf.\n\u003e \n\u003e https://zuul.opendev.org/t/openstack/build/048215afe33544209d4c5298e4b80b6a/log/logs/systemctl.txt#2489-2509\n\nok, i\u0027ve proposed several ways to maintain backwards compatibility in swift package, let\u0027s see the opinion from maintainers.","accounts_in_message":[],"_revision_number":1},{"id":"a20c4a6d7812fb4ba1baa0440168ef44fd602e4a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-03-24 15:53:29.000000000","message":"Abandoned\n\nreverting the change in packaging until a better solution can be handled by swift maintainers https://review.rdoproject.org/r/c/openstack/swift-distgit/+/32516","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"579c6774115d8afd647edab8ca73ef4bd39985d1","revisions":{"579c6774115d8afd647edab8ca73ef4bd39985d1":{"kind":"REWORK","_number":1,"created":"2021-03-24 09:40:11.000000000","uploader":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"ref":"refs/changes/76/782676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/76/782676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/782676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/782676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/76/782676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/76/782676/1"}}},"commit":{"parents":[{"commit":"8b42834b9a2347c4a4dff2c5840852d659218dc3","subject":"Update supported Ubuntu version to Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8b42834b9a2347c4a4dff2c5840852d659218dc3"}]}],"author":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-03-24 09:35:45.000000000","tz":60},"committer":{"name":"Alfredo Moralejo","email":"amoralej@redhat.com","date":"2021-03-24 09:40:09.000000000","tz":60},"subject":"Add object-expirer section to object-server.conf","message":"Add object-expirer section to object-server.conf\n\nSwift is moving object-expirer service to use object-server.conf config\nfile in [1]. This review is adding the required oject-expirer section\ninto this config file as required by the systemd service.\n\n[1] https://review.rdoproject.org/r/c/openstack/swift-distgit/+/32374\n\nChange-Id: I2ed6b5be63923b01e300e8bc07e716c60ab4751a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/579c6774115d8afd647edab8ca73ef4bd39985d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/579c6774115d8afd647edab8ca73ef4bd39985d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
