)]}'
{"id":"openstack%2Fpuppet-swift~783190","triplet_id":"openstack%2Fpuppet-swift~master~I61c2ced2935a88b4a86b0a9b8cc64efbbe027bd4","project":"openstack/puppet-swift","branch":"master","topic":"tripleo-undercloud","hashtags":[],"change_id":"I61c2ced2935a88b4a86b0a9b8cc64efbbe027bd4","subject":"Drop the tripleo-puppet-undercloud job","status":"MERGED","created":"2021-03-26 06:36:43.000000000","updated":"2021-03-31 02:08:13.000000000","submitted":"2021-03-31 02:08:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783190-tripleo-undercloud","meta_rev_id":"1201e810cb5f3f4387d380264f3348a6aed590c4","_number":783190,"virtual_id_number":783190,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 02:08:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-03-26 08:18:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-03-30 16:25:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-03-30 16:25:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 08:11:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 08:18:25.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-03-30 16:25:02.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8b60fa35befdf603c6123c8c45827f3cbee41ad8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 06:36:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a3995d4b97dcc350fc36404d729a61ab4d8385e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-26 06:37:10.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e0fe9216edeacfdbff200f40416ee38164f1abd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 08:11:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f2f21b0767504dc1a3c5d74b763b49b3 : SUCCESS in 4m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e47ed52c830d4a14a1ccf08414d04f39 : SUCCESS in 4m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8608449dc3b4007a339efc610a7df46 : SUCCESS in 12m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f5e4a834b0a04eb8a0e4956bfd075c16 : SUCCESS in 16m 16s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/85fce37c286944509773829e0c639671 : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47de0b31e7c843c18a8b61f871de1ef3 : SUCCESS in 47m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c4108ee3a12f4e109b4935f6cb8c3e7a : SUCCESS in 1h 18m 35s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f72bec7d63be4a36a7bff56e6ca6ebde : SUCCESS in 25m 39s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1faf7c67bfc47be9a303c2be1fc5b86 : FAILURE in 24m 46s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7efc2667668e45e0aa29d7289cd04965 : SUCCESS in 1h 29m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e5108127a49dfc7e6525d44fa0cb7c4e745014f3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-03-26 08:18:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ced53baeec70662bacdadb8bfe8db46b47a78210","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-03-30 16:25:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"15cbd3aa160b0fedfbc584c82bcfdef8d26f39e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 16:25:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dab6f226122d824d0d42222d5f0390d8748fb9e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 19:01:42.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/225193e891aa4cb29c052bcff4fc9e98 : SUCCESS in 3m 59s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef381a4705094ab5877959fb592b21fb : SUCCESS in 4m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e49b163d2f644528647378c5eebb1a7 : SUCCESS in 15m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fff8eca61e943ef8faf904a4e43a6e5 : SUCCESS in 14m 00s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c12574b46d5c4b67a19710e98812a17f : RETRY_LIMIT in 45m 15s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/473f24d8350a4d799e8d17f38e1be189 : SUCCESS in 31m 27s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d2926382a546088a03f03d733222ba : FAILURE in 21m 39s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9bce9558d7dc4eedbfd793eb32389630 : RETRY_LIMIT in 48m 04s","accounts_in_message":[],"_revision_number":2},{"id":"d80ea81307bcc5b5f0c323746148aa4faa5c53df","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-03-30 22:37:10.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f1bd16f29bf3c33f086504bf60728fea49a5fbe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:38:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c6a2c87bd19e465f8452291617e27bf4 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94d61ba9f2934576a33348cc75b89e21 : SUCCESS in 4m 01s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0925138ca61f40d1b1b985a2dd25a073 : SUCCESS in 12m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b16aba11536549cfae820049eca39985 : SUCCESS in 13m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d46f812141b42d5a9085eea76b3fd93 : SUCCESS in 12m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8afcd557b055429697047961d57efeab : SUCCESS in 52m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/638efd6a27b743e6982aed3dcb171aed : SUCCESS in 1h 16m 02s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/8039eb27acdf4fdf850ed49a96129eb9 : SUCCESS in 25m 52s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/38a76117d0984c0aaff7d78244270709 : FAILURE in 22m 50s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e66c48b17dd645f881b8153f306d2881 : SUCCESS in 1h 31m 11s","accounts_in_message":[],"_revision_number":2},{"id":"769c90c162f6509379aed43270e6a9df1be98d31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:38:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4c064c2b58f45c336d4e3909ef1ab7a76b7d9176","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:08:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c81849705784d48a5f7f362ce55b721 : SUCCESS in 4m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cd4beb18e464211b1c8978d64e7cb05 : SUCCESS in 4m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/73b5bc26a20a4cd198435229abe96bb1 : SUCCESS in 15m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52c2162c816b48f385fadd7d78ab3ff0 : SUCCESS in 13m 45s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ca2b314e3f04f80a571bcac7c900036 : SUCCESS in 1h 26m 07s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b09fa0da37b94f5cb894b9a8be851d4f : SUCCESS in 34m 18s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e6511321e2134789a05bdf0565f6fd13 : FAILURE in 20m 41s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b9f97d19dbb547fa80016a7c6ebe81d5 : SUCCESS in 1h 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"1201e810cb5f3f4387d380264f3348a6aed590c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 02:08:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"77f06130e3b769e1d11f6310b077a872c838e35d","revisions":{"d318dd0bd146f79752ef8335bf56224271effd2d":{"kind":"REWORK","_number":1,"created":"2021-03-26 06:36:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/783190/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/783190/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/1"}}},"commit":{"parents":[{"commit":"8b42834b9a2347c4a4dff2c5840852d659218dc3","subject":"Update supported Ubuntu version to Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8b42834b9a2347c4a4dff2c5840852d659218dc3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:35:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:35:48.000000000","tz":540},"subject":"Do not run tripleo-puppet-undercloud jobs","message":"Do not run tripleo-puppet-undercloud jobs\n\n... because Swift is no longer enabled in Undercloud by default.[1]\n\n[1] b3f4111c9ebb6ebd572b734a46b1d58e2875bd55\n\nChange-Id: I61c2ced2935a88b4a86b0a9b8cc64efbbe027bd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d318dd0bd146f79752ef8335bf56224271effd2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d318dd0bd146f79752ef8335bf56224271effd2d"}]},"branch":"refs/heads/master"},"77f06130e3b769e1d11f6310b077a872c838e35d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-26 06:37:10.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/90/783190/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/90/783190/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/90/783190/2"}}},"commit":{"parents":[{"commit":"8b42834b9a2347c4a4dff2c5840852d659218dc3","subject":"Update supported Ubuntu version to Focal","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/8b42834b9a2347c4a4dff2c5840852d659218dc3"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:35:48.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-03-26 06:36:56.000000000","tz":540},"subject":"Drop the tripleo-puppet-undercloud job","message":"Drop the tripleo-puppet-undercloud job\n\n... because Swift is no longer enabled in Undercloud by default.[1]\n\n[1] b3f4111c9ebb6ebd572b734a46b1d58e2875bd55\n\nChange-Id: I61c2ced2935a88b4a86b0a9b8cc64efbbe027bd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/77f06130e3b769e1d11f6310b077a872c838e35d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/77f06130e3b769e1d11f6310b077a872c838e35d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
