)]}'
{"id":"openstack%2Fpuppet-swift~784327","triplet_id":"openstack%2Fpuppet-swift~stable%2Fwallaby~I5f90de3c18a3cea2efd12b98f182ecb8d8d4c20e","project":"openstack/puppet-swift","branch":"stable/wallaby","topic":"create-wallaby","hashtags":[],"change_id":"I5f90de3c18a3cea2efd12b98f182ecb8d8d4c20e","subject":"Update .gitreview for stable/wallaby","status":"MERGED","created":"2021-04-01 09:36:29.000000000","updated":"2021-04-05 06:17:13.000000000","submitted":"2021-04-05 06:17:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784327-create-wallaby","meta_rev_id":"0bcff959537a22d5c78810e30bf00345b289139b","_number":784327,"virtual_id_number":784327,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-05 06:17:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-04-01 13:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-04-05 05:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 13:34:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-04-05 06:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba8f6f561cbf2228ab569d66c8d7180edef2acbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2021-04-01 09:36:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08169d3c8b1221f9f7792bd645d3b57325d7d310","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-01 13:34:50.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3eb9c178c9910df87e63915935702bc2b1d8d26a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-05 05:56:02.000000000","message":"Patch Set 1: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b24f2352f6a196437866e1bef944251272d8daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 06:02:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4960b022e02490ab83a66a41daa9241 : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":1},{"id":"139bd2320756b333c8cb272e6fc52a740567a262","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 06:06:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c7d4d093289382faa1e621249fa92d2c24485cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 06:17:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb602a6dfca048f18bbc5cc7da9404e0 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0bcff959537a22d5c78810e30bf00345b289139b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 06:17:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16ac3771a7f72206b4bce8f650ab009e9648b95f","revisions":{"16ac3771a7f72206b4bce8f650ab009e9648b95f":{"kind":"REWORK","_number":1,"created":"2021-04-01 09:36:29.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/27/784327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/27/784327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/784327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/784327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/27/784327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/27/784327/1"}}},"commit":{"parents":[{"commit":"492702370e78338b2053eda522636446df8e64be","subject":"Merge \"Drop the tripleo-puppet-undercloud job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/492702370e78338b2053eda522636446df8e64be"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-04-01 09:36:28.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2021-04-01 09:36:28.000000000","tz":0},"subject":"Update .gitreview for stable/wallaby","message":"Update .gitreview for stable/wallaby\n\nChange-Id: I5f90de3c18a3cea2efd12b98f182ecb8d8d4c20e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/16ac3771a7f72206b4bce8f650ab009e9648b95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/16ac3771a7f72206b4bce8f650ab009e9648b95f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
