)]}'
{"id":"openstack%2Fpuppet-swift~795832","triplet_id":"openstack%2Fpuppet-swift~master~I85abf3811d61fa8bfc0a1607818d6495549b5a6b","project":"openstack/puppet-swift","branch":"master","topic":"xinetd_as_arg","hashtags":[],"change_id":"I85abf3811d61fa8bfc0a1607818d6495549b5a6b","subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","status":"MERGED","created":"2021-06-10 15:18:59.000000000","updated":"2021-10-03 14:09:58.000000000","submitted":"2021-10-03 14:08:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"795832-xinetd_as_arg","meta_rev_id":"1169fcad178a12c06036133fa58ee558487a6098","_number":795832,"virtual_id_number":795832,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-10-03 14:08:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"recommended":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2021-10-01 07:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":2,"date":"2021-10-03 12:30:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-10-01 18:49:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-10-01 08:00:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"value":1,"date":"2021-10-03 12:30:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-10 16:14:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-11 10:35:52.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"},{"updated":"2021-06-11 10:35:52.000000000","updated_by":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"CC"},{"updated":"2021-06-14 20:52:15.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-06-15 11:17:53.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2021-06-15 11:32:20.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"state":"REVIEWER"},{"updated":"2021-06-15 11:37:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2021-06-15 11:37:19.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2021-06-15 11:49:07.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-07-06 09:52:57.000000000","updated_by":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"46ba1180c6c886d56f5626462661253d0e3d5665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-10 15:18:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9197fdce5f4d3badc9d9569935f7f2c061081a6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 16:14:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ab64afaefb44ab4a0cf8c55be931f11 : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/347e95a1b6944b7b98dda119ba2f8959 : FAILURE in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b7606a41cb840c08af0b2940050eceb : FAILURE in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/be0802aaa1fd476aae067675fba779ac : FAILURE in 18m 00s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/742df4dc2eb04365928f9ad542195632 : FAILURE in 18m 48s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde5ab36408c44cc8c733156b1f03a26 : FAILURE in 15m 03s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6936ec388134b398b12a560e65a5192 : FAILURE in 15m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cc0564dc20664c27ae2e0f581a2eaccf : FAILURE in 15m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95f09a502c5640af93f8d0433586365f : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4807fcb139594a858e6dc01e39855e93 : FAILURE in 15m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/43f3180cdfd048bba6508ef47f82ab00 : FAILURE in 15m 02s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/39ad91b77af841f0a36d3d64c47ffe7e : FAILURE in 12m 51s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc06f464f7934c1e88d7c67cc2d5bd00 : FAILURE in 16m 02s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9b0df62c3f54164b667f8901ce3457d : FAILURE in 11m 01s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27734238f0d1470a96b0e29f1bf3149f : FAILURE in 8m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24f112d55eb84607a46f0e31d6ad77d6 : FAILURE in 7m 50s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/847b0e52b7dd481a95b0ee09f80b927f : FAILURE in 53m 58s","accounts_in_message":[],"_revision_number":1},{"id":"35241841cb7b74f47dd674d55d1efac5fda4ac8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-10 16:56:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"19ced70c5c0a87e4a2acc37127cddac2d4dc5839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 18:53:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc3c960d39564b1b9fdb95143a384ee3 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9d3365c3f9f4961a8d17d798b816cf0 : SUCCESS in 4m 46s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1ddff0ff64948048d34d9d42551fd44 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9375db1624b94b39828561b3e8fdf30b : FAILURE in 10m 00s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/47a6b02d7db04007913e47bb42815bd8 : FAILURE in 10m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18d836f0079a4b9a98641ade46104009 : FAILURE in 7m 28s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ef564af389346b9b5dcf57f53a5ff10 : FAILURE in 8m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac8c8d11d3b3413b8695b29a7a5d69c0 : FAILURE in 7m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a40b1a1323634ea89a95ef81b53266bd : FAILURE in 37m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05559ced65704331ad0138b57d861e8e : SUCCESS in 1h 28m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/20ebe1aadc3d41b4bbb7ad6db32ad880 : FAILURE in 1h 48m 08s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/70e4dffef4a64fbc9b7359f4a810e5cb : SUCCESS in 29m 31s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/938b2aa640ba482e89a1413a6e1f2f02 : FAILURE in 25m 09s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/a01736834f7644b190bde43d7640cc65 : SUCCESS in 30m 48s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/183cc5d53ca448bcaf041db4dac51ff3 : FAILURE in 24m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/64da63b28b5c4ba98e64ceec9d0e30cf : FAILURE in 27m 34s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44f696a527784456baec75d5b70b156b : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":2},{"id":"5d720c33ec83dece5d7287123c184f4301205905","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 06:51:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"26a0a7cb9bb41fec6d98092895e5f290fdf5229b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 07:32:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3571c04dfafbd1c08000dd398372978b4958d7ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 08:11:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a945201b4aa5f24b084176f41dc5db9266beae56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 08:50:29.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ff078e3534e572ba21c5683d5ec0271457971dbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 10:28:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa4a81fc184449c68856e518ae018750 : SUCCESS in 4m 26s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1fdd160877c42718c0c549714eb2a34 : SUCCESS in 4m 42s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69b418a569ab4c1faa993839c82e9124 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f73280c7e9594567bb6cce126a288f37 : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bee1c74a60bf4e338cf06a44c854f846 : SUCCESS in 16m 47s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c456db96084440c1a42997de4e5f9b29 : SUCCESS in 15m 04s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/925198048b7643bdb2ad2cf889fb51c2 : SUCCESS in 16m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e750040a27af417a91d0260d1c88b34a : SUCCESS in 13m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42534b65c0b84f14a89e2917510e2ee9 : FAILURE in 40m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/072c84951d534779b64b47e3fb0772ee : SUCCESS in 1h 26m 23s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db5f8fdcbcb84b32816002cbf6ceaff8 : SUCCESS in 1h 32m 58s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d243cc2dccd84108b8e597e1a0112c53 : SUCCESS in 33m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c554c9878ac493cbfa738318864a021 : FAILURE in 26m 15s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6c7d7f45b79047b7a0dd5952d5700c47 : SUCCESS in 29m 50s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e977334553dc4565b60e0e4f13bf47de : FAILURE in 22m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3631a77a9c9e41839c51e47263ab7e59 : FAILURE in 26m 05s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ee4ebabb092485a8cf0a9e54a5a09a3 : SUCCESS in 1h 27m 09s","accounts_in_message":[],"_revision_number":6},{"id":"59b65372388bd2871a1e6a39746b87a9b3b1fffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 10:28:51.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"871e30d57457313b1be29fcb7884b543f77e30ee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 10:35:09.000000000","message":"Topic xinetd_as_arg removed","accounts_in_message":[],"_revision_number":7},{"id":"3ecb3ce70aec006cf569b3a509c15c09f34dc55b","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-06-11 11:13:19.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"79dfa38cbb8ccbc893bbca01b5410ff189950a84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 12:12:57.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"291cda18d78098c4e25343b717c3919f7541d61e","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-06-11 12:22:24.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6cc8eeb57a82df4a854a9e8c5c82852fc18b8265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 13:46:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7b94f2ff75174f478a6fb106c88d5f3e : SUCCESS in 5m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba78ad1bad354c18a3167efaf36fafdd : SUCCESS in 4m 03s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaa3c1188f704ffc93a773aa77867ab5 : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/42629ef708d54c108bf18cccae58c764 : SUCCESS in 14m 34s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/37fa9d746dd14d34932c7c5e64413dca : SUCCESS in 14m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc1c34ff727b4c209a81a93b9ce0b301 : SUCCESS in 12m 52s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8fe7163b6f354bb4abf7a6ab75eaa0df : SUCCESS in 13m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34a0fe49072444ab802c6242c9ec874b : SUCCESS in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36f6998a5ab34682a05b3c89e40a7e6b : FAILURE in 40m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/017712d0c63c46099891914f15ffa1da : SUCCESS in 1h 21m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7ce445f92dfc433191de164b68db7f28 : SUCCESS in 1h 31m 41s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c075373bed04e10a07af4e19ff9d789 : SUCCESS in 31m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36082fc156364eff92c9b5740e0641bb : FAILURE in 24m 15s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc7487176ae74d01805c68e010e50fce : SUCCESS in 28m 18s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796e2b8901d34de0bb06b4ddf3b3cb6b : FAILURE in 26m 47s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9323c04000e34914972f1f73214919c0 : FAILURE in 24m 42s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a9e31806f273441b810d535da588991a : SUCCESS in 1h 24m 01s","accounts_in_message":[],"_revision_number":8},{"id":"624dceb444cc0d35fe61e96ae28b8469bb67b6a6","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-06-14 20:52:15.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"406f7ddca4eda49eaba4cccb84a0d7dd25cae9a7","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2021-06-15 11:17:53.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"704cbe3610222a169e7ba7837dbe92373a4e2298","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-06-15 11:32:20.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"6333774d987a1d2b876702dda3b222fbe7d71596","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2021-06-15 11:37:19.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"9dca2c5f89845ac0cb691eed6b2fc1599f1df499","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-06-15 11:39:10.000000000","message":"Patch Set 8: -Code-Review\n\nI see swift::storage is used from swift::storage::all:\n\nhttps://github.com/openstack/puppet-swift/blob/master/manifests/storage/all.pp#L171-L173\n\nBut in all.pp does not include the option to enable/disable xinetd, i guess it should be added there too?","accounts_in_message":[],"_revision_number":8},{"id":"a7bca3a61f58b08000b6bc3b22eb1b52cc9fd42f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 11:49:07.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)\n\n\u003e Patch Set 8: -Code-Review\n\u003e \n\u003e I see swift::storage is used from swift::storage::all:\n\u003e \n\u003e https://github.com/openstack/puppet-swift/blob/master/manifests/storage/all.pp#L171-L173\n\u003e \n\u003e But in all.pp does not include the option to enable/disable xinetd, i guess it should be added there too?\n\n+1","accounts_in_message":[],"_revision_number":8},{"id":"b830497246a7401e6d544557dd44a01f10ede1e0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 11:50:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"5936bfab7741ee389e5e0d19c0dc2ef0cfacfb8f","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-06-15 11:53:42.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fe68428034c00db8eb9eeed8c120d21e8b847be2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-15 12:13:08.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bff203accd32b1e1856ac62f00ac75b4dd8d67e9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-06-22 14:55:36.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"e31775a53fe7f2758a167d0720005ae6167de8b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-28 09:29:08.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"9ae11b9071675f98c202befd053ca70ff1182cb9","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-28 09:31:12.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"34b43d1e577375cc5b3ccf42e4673c00b5686142","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-28 09:39:49.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"c1c62e0b69d210d9831c72c992efac17dfb6b7a6","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-09-28 11:02:33.000000000","message":"Patch Set 9: Code-Review-1\n\nswift::storage has $rsync_use_xinetd parameter:\n\nhttps://github.com/openstack/puppet-swift/blob/master/manifests/storage.pp#L28\n\nBut when it\u0027s called from swift::storage::all it\u0027s not configurable, what will lead to issues in CS9:\n\nhttps://github.com/openstack/puppet-swift/blob/master/manifests/storage/all.pp#L171\n\nWe should be able to set the parameter from ::all or maybe make the default value for rsync_use_xinetd to switch to false for CS9 (or both).","accounts_in_message":[],"_revision_number":9},{"id":"53f8b5f02b009ee64055eb208281cca2522da963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-28 11:07:07.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff8098e43b6f4c89a94b03f010cb4a60 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e24cf4dd40848bca574e487fb054bd7 : SUCCESS in 4m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0bf3519c7db6419888e59963fd88cd72 : SUCCESS in 4m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfe2dd705f0c4c2990a96352e9d7a355 : FAILURE in 16m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a1f7d35c497a43b586a8fff51f5eb666 : FAILURE in 15m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2aad1ea305764f47a3c11ae35094daf3 : FAILURE in 15m 40s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fba06b33f8e04dcf90581b2a046d145f : FAILURE in 13m 26s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ae45bb3066f4b4baa51b54f9079eac3 : FAILURE in 14m 16s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f13b287c05949ce8269a66c0239b918 : SUCCESS in 52m 11s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/27f7100fa5f444c2ad1b12fcfe4ea87d : SUCCESS in 1h 32m 48s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d09a14da0ae48e4894e2dc0e12e4b75 : FAILURE in 22m 33s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/9c3b3163d631452aa85203f146c62288 : SUCCESS in 31m 15s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d38d1d1c75ea4abb970535cc68e5b030 : FAILURE in 24m 16s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/789649a9da2446eea3151173abcfecf8 : FAILURE in 24m 41s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9e5c6827777744c5856ed07524462290 : FAILURE in 44m 42s","accounts_in_message":[],"_revision_number":9},{"id":"17cadd43839b60590486cd694e13878cd5484963","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-28 12:57:14.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9: Code-Review-1\n\u003e \n\u003e swift::storage has $rsync_use_xinetd parameter:\n\u003e \n\u003e https://github.com/openstack/puppet-swift/blob/master/manifests/storage.pp#L28\n\u003e \n\u003e But when it\u0027s called from swift::storage::all it\u0027s not configurable, what will lead to issues in CS9:\n\u003e \n\u003e https://github.com/openstack/puppet-swift/blob/master/manifests/storage/all.pp#L171\n\nAlthough we can override the parameter outside of swift::storage::all (by defining defaults), it makes sense to me to implement an interface to swift::storage::all as well for easy switching.\n\n\u003e We should be able to set the parameter from ::all or maybe make the default value for rsync_use_xinetd to switch to false for CS9 (or both).\n\nI think you can follow what I proposed here about this.\n https://review.opendev.org/c/openstack/puppet-ironic/+/797668/37/manifests/params.pp\nBasically what I\u0027d suggest is to implement a parameter to represent availability of xinetd in the base params class and use the parameter to define default value and validate user input.","accounts_in_message":[],"_revision_number":9},{"id":"049241c23bc30b08c03a6d83b48b0c64793bed73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-30 09:55:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"817ab1312a6c2df6fea8ecbbf3e4c88db32684cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-30 09:57:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"25e6af1e05152126728d48d0837f7c944bb43598","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-30 10:19:01.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"59368efd129a48d1fc2e94f78f1a306c0e12a65e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-30 11:05:57.000000000","message":"Patch Set 12: Code-Review-1\n\n(4 comments)\n\nAlso, please add a release note to explain the new parameters.","accounts_in_message":[],"_revision_number":12},{"id":"932f44e04ab365410265ac236daf81d06cd89105","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 11:59:12.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d811dbd5eaf4f52929fc0b8139bdb1a : SUCCESS in 4m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99b03de4056c45dfb95b3a3ad8225903 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8116f686f3a1458784d1fade8f1a3129 : SUCCESS in 3m 43s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1915acb9fd974ff6a45d35a5a16e3c4f : FAILURE in 17m 06s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3862353ab7e24423963218b2982b0d75 : FAILURE in 17m 16s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8446654fc93d438091660f05c7ccd8bb : FAILURE in 17m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a7a9e584af14166af3050d014d835f2 : FAILURE in 14m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e44ba9d6f1a048e1b238fa843c93b52a : FAILURE in 14m 36s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa27c8f85cff4f4db1fbcfa1e76070e9 : FAILURE in 8m 15s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c860ebfa95134bac99dc02b28f84c867 : SUCCESS in 1h 33m 26s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a420a12a7d62466da04ec4dfdfe9197d : FAILURE in 24m 27s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac7c3390a23b4ef0b09bdf54f1765d48 : SUCCESS in 31m 49s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66b81bb7b2ca455a9c355d5effbc24fb : FAILURE in 25m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0eba5ce8de7d4c78b22b51ed78736991 : FAILURE in 28m 03s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c87adcf42e5e4d2689c57679b057fe46 : FAILURE in 47m 06s","accounts_in_message":[],"_revision_number":12},{"id":"1619e91a23eceb91fb795f0336342b5a0c953d9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-09-30 12:44:29.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"07c32cf0c6bb9eb24dede0a2139a5bfad24d5107","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 14:19:06.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e6def282d5c4ea79c25fa9fc85abb2b : SUCCESS in 3m 48s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e517fb3a0de44b987b50320efb2d359 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ee6a8ed510547f19f2ef3afbecc8e29 : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a917f09e271493d9e84a36e2286fd12 : FAILURE in 16m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0eb75bcd8e4649bfa5ef13fc380d8e54 : FAILURE in 16m 10s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c9339d374242468a9ed782317affab5b : FAILURE in 14m 19s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1efae25bd56949ea8bf3523f941e4bcd : FAILURE in 13m 15s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb7469ab8d29437398fafe1284a61c87 : FAILURE in 12m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57d7b40174cd429d99fbf5adac8aced9 : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/02dd67c6b0bb47db9412f5bfc11d7917 : SUCCESS in 1h 33m 17s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09be6b1b63fe4072849d8f39e96b24b0 : FAILURE in 23m 35s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/eed36579de514920863e776df6283850 : SUCCESS in 29m 36s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c8d88f7dca9419cb9d48c351b6c1161 : FAILURE in 25m 22s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98eaa420e8c94c7ca7e1553e03eec164 : FAILURE in 23m 23s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/194ea408deb54d0592df47251787fb0f : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/0cadb80fd021462baa365a2a319e7baf : FAILURE in 47m 05s","accounts_in_message":[],"_revision_number":13},{"id":"ec39e8727c07bc4bc5e7d46603180f8e7cad0854","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-10-01 06:40:25.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 12: Code-Review-1\n\u003e \n\u003e (4 comments)\n\u003e \n\u003e Also, please add a release note to explain the new parameters.\n\nThank you Takashi for the assistance. Now we\u0027are hitting an issue which does not seem related to the change after deep investigation. Do you have an idea of what could be the reason of \n\u0027expected that the catalogue would contain Package[xfsprogs] with ensure set to \"present\" but it is set to \"installed\"\u0027 ?","accounts_in_message":[],"_revision_number":13},{"id":"53d03253ed98cc7fa95eb56ea1aff0b9ad6b057a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 07:05:30.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"9c59a6357941174fbb4040138ca51aba2297e47e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 07:13:34.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"26dec11a04f5724041fb5d03465fb9a53941f181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 07:16:40.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"fdab6dd23d0c6279bbb2d80b6490bbb88a024383","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 07:18:28.000000000","message":"Patch Set 16:\n\n(1 comment)\n\nGood catch. That error seems to be the known issue caused by recent bump of puppetlabs-stdlib. I have submitted a fix for it(which is now the parent patch of this) and let\u0027s see whether it works.\n\nI also made some minor updates with adding validation to prevent usage of xinetd in CentOS9. Hope all look good to you.","accounts_in_message":[],"_revision_number":16},{"id":"8f3096725046e412a1fc75334d1e5d831150aff9","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-01 07:24:39.000000000","message":"Patch Set 16: Code-Review+1\n\nThanks Takashi for your help!","accounts_in_message":[],"_revision_number":16},{"id":"8335e497e46a1d94855d72a626933406b263d9b5","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2021-10-01 08:00:10.000000000","message":"Patch Set 16: Code-Review+1\n\n\u003e Patch Set 16:\n\u003e \n\u003e (1 comment)\n\u003e \n\u003e Good catch. That error seems to be the known issue caused by recent bump of puppetlabs-stdlib. I have submitted a fix for it(which is now the parent patch of this) and let\u0027s see whether it works.\n\u003e \n\u003e I also made some minor updates with adding validation to prevent usage of xinetd in CentOS9. Hope all look good to you.\n\nAll good for the changes added, thank you","accounts_in_message":[],"_revision_number":16},{"id":"5568a300bcf8fbb3d5b7090eecbee8e367782140","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 08:38:08.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e13dfdd3c3934a56885a8cc334cbe7df : SUCCESS in 6m 00s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/304cc2fa57e643f09e4aa650bfbab53a : SUCCESS in 6m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c60f77337744f37881744d8f18d58cc : SUCCESS in 5m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/31a5f275d8bb47019daa69c93e9676ed : SUCCESS in 16m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb8fc6d88a964e108d5b747d6f6190fb : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d3ed59f39bb7464590e50ada0f3c85c9 : SUCCESS in 17m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed95033cf55e438cbb12601b283da3b7 : SUCCESS in 15m 07s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3d6a6f7541a46c69a450082252fe679 : SUCCESS in 17m 41s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3d487859ed0484299925d5d99791370 : FAILURE in 12m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a9fd6281bed64b7c856a7583729605a4 : SUCCESS in 1h 18m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f75bf903ace24050a77b0dd434ee39dc : FAILURE in 30m 05s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a56f55a59e5543e38c9cb045f5989135 : SUCCESS in 31m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e34b683555041acb57ddd544b4592e1 : FAILURE in 27m 28s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c306276799e4b3f96fd00c621fe6562 : FAILURE in 26m 33s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d0b847a2f3c4b38a7eb486b1245a664 : SUCCESS in 4m 31s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/55b04d89ffe64ee5b1bc2c17579eccbe : FAILURE in 45m 41s","accounts_in_message":[],"_revision_number":16},{"id":"5a20727ca49e186a04d8177a051ea05a227d51a3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-01 09:26:01.000000000","message":"Patch Set 16:\n\nThere is some problem with latest promotion in tripleo. Fripn #oooq channel:\n\n\u003carxcruz|ruck\u003e amoralej: not a problem with the job, the problem was the toolbox, it was without dns (don\u0027t know why) and so it wasn\u0027t able to tag it properly","accounts_in_message":[],"_revision_number":16},{"id":"37bb0cc5aa68cf60f49ab0632830a92e2028ff2a","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-01 13:43:59.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"aa146d1a2db8063334c2f0842b19159d61a062c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 15:54:11.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de7c6ec8a9442f9ad5b800029ecfa23 : SUCCESS in 4m 29s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7049971eae8d470daeb156a01bff0f87 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8b37d56bcf544dea61c701d5e276cca : SUCCESS in 3m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/54be1de081d44ab09ed93e741cf0cb46 : SUCCESS in 13m 55s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7c1bcd7f0914a6f8bfb34c54d14f865 : SUCCESS in 15m 49s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72191a7f398b4aca80177b774b2eac7a : SUCCESS in 12m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7e3fe856859474c89c31776cf790b5e : SUCCESS in 14m 14s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c06a34e3b06147789d32fa7f222257ba : SUCCESS in 15m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f676e63b4a44d5b8cd057ec366f9adc : FAILURE in 7m 24s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7a9637572eec4310b8263ac700c80b3e : TIMED_OUT in 2h 04m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fedb5de17ee4ab0b9e89ec1ae4a4340 : FAILURE in 28m 55s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d38be59020384269ad178ee1dc65c6a8 : SUCCESS in 29m 03s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5bca7740a33457082e0ca1dad2b2a5a : FAILURE in 23m 43s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20f6a4b8b7a44e9eb9a64049292ce448 : FAILURE in 27m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/08b060d8e08b49909222a66dd50dd2c8 : SUCCESS in 4m 22s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/684fd55879994c35ac6d5900bbf5a3b7 : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":16},{"id":"b0e7ec163f19076eb3fcf3f3d01508c8c0c3fca3","author":{"_account_id":16312,"name":"Alfredo Moralejo","email":"amoralej@redhat.com","username":"amoralej"},"date":"2021-10-01 16:04:53.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"1cda2574afe6084f0f80866c3f7082fecbe8a1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:45:03.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cdf91e9833ae44e4b08b61c50e815d20 : SUCCESS in 4m 23s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9575779feb854d218e19908fcd9c243a : SUCCESS in 4m 52s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c366de92611c4e63937bf051b9f8ad35 : SUCCESS in 4m 10s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/25be3ba0c22d4cd996067995a58d68ea : SUCCESS in 17m 24s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf6b51c82c74488fbb8e58604052deef : SUCCESS in 15m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb4382a686064b66bf4139336bf2ed2e : SUCCESS in 16m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43cdcfde8c3e465c97f3176983ccb679 : SUCCESS in 16m 36s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/730d190cc83d4dc29b64678f97e36495 : SUCCESS in 15m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de5e54b48ef940c2b5c658a64c2ac74f : FAILURE in 8m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0a3d9e946ad41d29055bc1d1b1913e6 : SUCCESS in 1h 25m 05s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3db58f9f84814bb791cb9c73ec5f80a6 : FAILURE in 27m 13s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2d9b0313bc24e83b191b372fa298891 : SUCCESS in 31m 19s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7109aa114560452b9d17ea6eb34defb3 : FAILURE in 24m 13s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1f2e2bb0125642af88b4312c602c426d : FAILURE in 29m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c06d9caed5ad4f8fa7cb64dcd668d330 : SUCCESS in 4m 26s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/38ff92a003fd4fe3bc3e98160371c397 : SUCCESS in 1h 33m 59s","accounts_in_message":[],"_revision_number":16},{"id":"9139d60179bc9d089f9b9c30851d6b2633356035","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-01 18:49:58.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"91ab5839f0c8af506f03d4a267439eaff6b5e2f3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-03 12:30:45.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"0365c2c2956cfe412a64728ed642540cb90834f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 12:31:51.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e961ba9246ce8ac6cd67c8a138749737856cfd17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 14:08:42.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21ceadf1712f419b80f9e076b0b5c245 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/549001edeeda45d9989f19813bf40d3e : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0885123680104e3d9d7a93c0392f4faa : SUCCESS in 4m 21s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d41c770d39374bebac1e60c9ca424d52 : SUCCESS in 15m 47s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/733edd7269c74affafd83c55495c1f7f : SUCCESS in 14m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad69aa918c0c4ce5a8356b5f7675b264 : SUCCESS in 13m 09s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f27de268630246a596f569efb103b827 : SUCCESS in 15m 09s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/289b0d94a5c0445d8540ff9769822c3b : SUCCESS in 1h 23m 28s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a86824f1dadb4bf38a304d859fe631d1 : SUCCESS in 30m 01s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd23a43b32744868863735e5ed740576 : FAILURE in 24m 41s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c6a7329891040dcaa6bdeddf356eb66 : SUCCESS in 4m 00s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/2df7ae9fe5744a929463ba19452c617c : SUCCESS in 1h 31m 43s","accounts_in_message":[],"_revision_number":16},{"id":"8d2164ed2982e3b8ed51cba84b2c7f8ad3e91bd1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 14:08:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"1169fcad178a12c06036133fa58ee558487a6098","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-03 14:09:58.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c5d319b47874571bc143ab1a0fcd89a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"053a3a2a56723e0ec43d1855ac7615657a4de8f0","revisions":{"0b920160961dcc5f6a12fff0ceec002214fb93a4":{"kind":"REWORK","_number":1,"created":"2021-06-10 15:18:59.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/1"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-10 15:18:58.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0b920160961dcc5f6a12fff0ceec002214fb93a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0b920160961dcc5f6a12fff0ceec002214fb93a4"}]},"branch":"refs/heads/master"},"1936b72adfb5d71a18d90677594a94481748e232":{"kind":"REWORK","_number":2,"created":"2021-06-10 16:56:37.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/2"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-10 16:54:06.000000000","tz":120},"subject":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1936b72adfb5d71a18d90677594a94481748e232"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1936b72adfb5d71a18d90677594a94481748e232"}]},"branch":"refs/heads/master"},"cf1741212d48a6003460afe6008fe8c4e53344b1":{"kind":"REWORK","_number":3,"created":"2021-06-11 06:51:59.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/3"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 06:50:32.000000000","tz":120},"subject":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf1741212d48a6003460afe6008fe8c4e53344b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cf1741212d48a6003460afe6008fe8c4e53344b1"}]},"branch":"refs/heads/master"},"c8228021456e2b4650243272c55efaccbff9d8e8":{"kind":"REWORK","_number":4,"created":"2021-06-11 07:32:12.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/4"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 07:31:39.000000000","tz":120},"subject":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c8228021456e2b4650243272c55efaccbff9d8e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c8228021456e2b4650243272c55efaccbff9d8e8"}]},"branch":"refs/heads/master"},"ee9eb12f8356669465744b1e794901e9557f6b8f":{"kind":"REWORK","_number":5,"created":"2021-06-11 08:11:07.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/5"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 08:10:46.000000000","tz":120},"subject":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ee9eb12f8356669465744b1e794901e9557f6b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ee9eb12f8356669465744b1e794901e9557f6b8f"}]},"branch":"refs/heads/master"},"a24da59b8d354affe35ac2438a35617104c7e896":{"kind":"REWORK","_number":6,"created":"2021-06-11 08:50:29.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/6"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 08:49:49.000000000","tz":120},"subject":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"WIP Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nCentOS \u003e\u003d 8.\n\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a24da59b8d354affe35ac2438a35617104c7e896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a24da59b8d354affe35ac2438a35617104c7e896"}]},"branch":"refs/heads/master"},"5c1b14969aafd24403cb3411d8ffeba23ce92ef9":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-06-11 10:28:51.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/7"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 09:10:09.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in RHEL/CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince RHEL/CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nRHEL/CentOS \u003e\u003d 8.\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5c1b14969aafd24403cb3411d8ffeba23ce92ef9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5c1b14969aafd24403cb3411d8ffeba23ce92ef9"}]},"branch":"refs/heads/master"},"e6a4b047d282f76dedeb884a38fb3613f781cd07":{"kind":"REWORK","_number":8,"created":"2021-06-11 12:12:57.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/8"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-11 12:12:27.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nPlus, we add a dependency on rsync-daemon in RHEL/CentOS \u003e\u003d 8\nwhen not using xinetd.\nSince RHEL/CentOS 8, rsyncd systemd unit is shipped in a\ndifferent package, rsync-dameon.\nWhen setting \u0027rsync_use_xinetd\u0027 to false in swift::ringserver\nand/or swift::storage, this package should be installed for\nRHEL/CentOS \u003e\u003d 8.\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6a4b047d282f76dedeb884a38fb3613f781cd07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e6a4b047d282f76dedeb884a38fb3613f781cd07"}]},"branch":"refs/heads/master"},"0b7fc0b2eafff31a6a73f27330e5df2bec4db67b":{"kind":"REWORK","_number":9,"created":"2021-09-28 09:29:08.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/9"}}},"commit":{"parents":[{"commit":"3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e","subject":"Merge \"Set fixture paths for unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3d292c53afcc8d877a4e0a2f0cf9b4f991222d3e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-09-28 09:28:13.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0b7fc0b2eafff31a6a73f27330e5df2bec4db67b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/0b7fc0b2eafff31a6a73f27330e5df2bec4db67b"}]},"branch":"refs/heads/master"},"97ede13feb817e3d0c69800bda2711e69aee078f":{"kind":"REWORK","_number":10,"created":"2021-09-30 09:55:49.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/10"}}},"commit":{"parents":[{"commit":"9f037d9f90d0e4f8d339c0470609a9afb5ab621e","subject":"Add license information in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9f037d9f90d0e4f8d339c0470609a9afb5ab621e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-09-30 09:55:37.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/97ede13feb817e3d0c69800bda2711e69aee078f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/97ede13feb817e3d0c69800bda2711e69aee078f"}]},"branch":"refs/heads/master"},"86c20cd6a996a0edefdeca01e07ca9d3dee25999":{"kind":"REWORK","_number":11,"created":"2021-09-30 09:57:50.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/11"}}},"commit":{"parents":[{"commit":"9f037d9f90d0e4f8d339c0470609a9afb5ab621e","subject":"Add license information in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9f037d9f90d0e4f8d339c0470609a9afb5ab621e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-09-30 09:57:46.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/86c20cd6a996a0edefdeca01e07ca9d3dee25999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/86c20cd6a996a0edefdeca01e07ca9d3dee25999"}]},"branch":"refs/heads/master"},"230596cb6fb09ef7f9cbf70ed010c48a184450b2":{"kind":"REWORK","_number":12,"created":"2021-09-30 10:19:01.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/12"}}},"commit":{"parents":[{"commit":"9f037d9f90d0e4f8d339c0470609a9afb5ab621e","subject":"Add license information in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9f037d9f90d0e4f8d339c0470609a9afb5ab621e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-09-30 10:18:32.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/230596cb6fb09ef7f9cbf70ed010c48a184450b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/230596cb6fb09ef7f9cbf70ed010c48a184450b2"}]},"branch":"refs/heads/master"},"a9904d79e8dbf3b0a6f87d29597a33f9b432deb4":{"kind":"REWORK","_number":13,"created":"2021-09-30 12:44:29.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/32/795832/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/13"}}},"commit":{"parents":[{"commit":"9f037d9f90d0e4f8d339c0470609a9afb5ab621e","subject":"Add license information in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9f037d9f90d0e4f8d339c0470609a9afb5ab621e"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-09-30 12:44:10.000000000","tz":120},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a9904d79e8dbf3b0a6f87d29597a33f9b432deb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a9904d79e8dbf3b0a6f87d29597a33f9b432deb4"}]},"branch":"refs/heads/master"},"fd45ebd23d36f5b607dd8ac3ec1bc6266082ff04":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-10-01 07:05:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/795832/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/14"}}},"commit":{"parents":[{"commit":"5f57e5521a04ebd08fa68d6d7671528a535c14bb","subject":"Fix unit tests incompatible with puppetlab-stdlib 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5f57e5521a04ebd08fa68d6d7671528a535c14bb"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-01 07:05:30.000000000","tz":0},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd45ebd23d36f5b607dd8ac3ec1bc6266082ff04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/fd45ebd23d36f5b607dd8ac3ec1bc6266082ff04"}]},"branch":"refs/heads/master","description":"Rebase"},"73c62d02e8489aa67c139d3fd55b275dddb80eb6":{"kind":"REWORK","_number":15,"created":"2021-10-01 07:13:34.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/795832/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/15"}}},"commit":{"parents":[{"commit":"5f57e5521a04ebd08fa68d6d7671528a535c14bb","subject":"Fix unit tests incompatible with puppetlab-stdlib 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5f57e5521a04ebd08fa68d6d7671528a535c14bb"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-01 07:13:04.000000000","tz":540},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/73c62d02e8489aa67c139d3fd55b275dddb80eb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/73c62d02e8489aa67c139d3fd55b275dddb80eb6"}]},"branch":"refs/heads/master"},"053a3a2a56723e0ec43d1855ac7615657a4de8f0":{"kind":"REWORK","_number":16,"created":"2021-10-01 07:16:40.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/795832/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/32/795832/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/32/795832/16"}}},"commit":{"parents":[{"commit":"5f57e5521a04ebd08fa68d6d7671528a535c14bb","subject":"Fix unit tests incompatible with puppetlab-stdlib 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5f57e5521a04ebd08fa68d6d7671528a535c14bb"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2021-06-08 15:25:31.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-01 07:16:19.000000000","tz":540},"subject":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver","message":"Add \u0027rsync_use_xinetd\u0027 as argument to swift::ringserver\n\nSince in RHEL/CentOS \u003e\u003d 8 rsyncd systemd unit is shipped in a\ndifferent package (rsync-dameon), we need to be able to not\nuse xinetd to manage rsync service.\nThe dependency on rsync-daemon for RHEL/CentOS \u003e\u003d 8\nis handled in puppet-rsync with [1].\n\n[1] https://github.com/puppetlabs/puppetlabs-rsync/pull/139/files\n\nCloses-Bug: #1930855\nChange-Id: I85abf3811d61fa8bfc0a1607818d6495549b5a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/053a3a2a56723e0ec43d1855ac7615657a4de8f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/053a3a2a56723e0ec43d1855ac7615657a4de8f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
