)]}'
{"id":"openstack%2Fpuppet-swift~796175","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~If2601644bbb8eeba3c103b25184e96e3c6f967ac","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","hashtags":[],"change_id":"If2601644bbb8eeba3c103b25184e96e3c6f967ac","subject":"Set fixture paths for unit tests","status":"MERGED","created":"2021-06-13 02:48:18.000000000","updated":"2021-06-28 14:16:29.000000000","submitted":"2021-06-28 14:16:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796175-bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","cherry_pick_of_change":794921,"cherry_pick_of_patch_set":1,"meta_rev_id":"fba5802c960d5088d6f76d61de4bc835ea6ff8f1","_number":796175,"virtual_id_number":796175,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 14:16:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-06-28 12:59:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-06-28 12:59:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-13 02:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ea504f93af8ec5e44daf07e5c59ad645b865b4b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:48:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"e7281d3753fad5248d4ae6bb48b80147bebf4540","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-13 02:50:25.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job tripleo-puppet-ci-centos-8-undercloud-containers depends on tripleo-ci-centos-8-content-provider which was not run.","accounts_in_message":[],"_revision_number":1},{"id":"8c9d8b80bc8fe969aefc68428d690fbd7d0df17f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:56:45.000000000","message":"Topic cherrypick-bug/1930403-stable/wallaby-stable/victoria-sluuvw5e7s removed","accounts_in_message":[],"_revision_number":1},{"id":"e18dbe44dde98977f57071fc14be09f34efe72e8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-13 02:56:48.000000000","message":"Topic set to bug/1930403-stable/wallaby-stable/victoria-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"2d1bbfa2453e402a5e417dde16393ea7e7b5cc2f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-06-28 12:59:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ba5ecfae26090e2f09e2fd8a55459c88351fabe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:42:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/75f749dcf20d46a6943913f26460dedc : SUCCESS in 6m 06s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06b55bab48e84620857c4a7a89fe3edd : SUCCESS in 6m 23s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/211168d8a2594ff4951a9c828b1bae61 : SUCCESS in 5m 54s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ec4e10f008d4c0c8115d370f5b2f3db : SUCCESS in 19m 00s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/537f68bc65d14288ad7dfac0e46e05b0 : SUCCESS in 19m 25s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d40c2caf998742e883223b2b3500ecc3 : SUCCESS in 12m 10s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5c4ee41ff45349ed84a8f8267de5b8e2 : SUCCESS in 17m 04s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b0c03442ef1c4c90a5fed0ac97f9db32 : SUCCESS in 32m 56s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/c22bc5e799b9462b8d2c0f5485faa456 : FAILURE in 35m 18s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65caf5f3afa9472198ac824d2143a148 : SUCCESS in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2897ad11f7e2b2f4f0dbc22b9daca061ba01710e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 13:43:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69219dca5e6355ce10c08d223b2fbb49408450d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:16:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ea78515e90642c0a71dd4c3a017cf01 : SUCCESS in 5m 23s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0f51d15e0b594d698e21833fa219a372 : SUCCESS in 6m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/490f60c433ec4be287fb3b070e341223 : SUCCESS in 6m 02s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/057caeec49c547a48064b2a09ddeb9f6 : SUCCESS in 16m 54s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/f95e606ba9ee4ef2a906d29eecd1d7e6 : SUCCESS in 15m 30s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f06022a1e2f246a29b9c4bfca0f8b789 : SUCCESS in 10m 53s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/821372d0a4ef427c87e56b736b91ca8a : SUCCESS in 31m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fba5802c960d5088d6f76d61de4bc835ea6ff8f1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:16:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b267de322a09b234d60c4e56ff82262779219b5","revisions":{"2b267de322a09b234d60c4e56ff82262779219b5":{"kind":"REWORK","_number":1,"created":"2021-06-13 02:48:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/75/796175/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/75/796175/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/75/796175/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/75/796175/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/75/796175/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/75/796175/1"}}},"commit":{"parents":[{"commit":"4f63aa2f109d578f40794a56e711b518b9ec52ae","subject":"Add TLS options for swift memcached client","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4f63aa2f109d578f40794a56e711b518b9ec52ae"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-02 05:24:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-06-13 02:48:18.000000000","tz":0},"subject":"Set fixture paths for unit tests","message":"Set fixture paths for unit tests\n\nThis change defines manifest_dir and module_path expicitly in unit\ntests so that modules installed under fixtures directory is properly\nloaded.\n\nCloses-Bug: #1930403\nChange-Id: If2601644bbb8eeba3c103b25184e96e3c6f967ac\n(cherry picked from commit 4eaee6777d27667e31f1ee013971d64a378fb33f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2b267de322a09b234d60c4e56ff82262779219b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2b267de322a09b234d60c4e56ff82262779219b5"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
