)]}'
{"id":"openstack%2Fpuppet-swift~800648","triplet_id":"openstack%2Fpuppet-swift~stable%2Fvictoria~Ia28f458ac84e4f46dcda0dd800276bbd5721a568","project":"openstack/puppet-swift","branch":"stable/victoria","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"Ia28f458ac84e4f46dcda0dd800276bbd5721a568","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-07-14 15:37:03.000000000","updated":"2021-07-20 20:39:04.000000000","submitted":"2021-07-20 20:39:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"800648-tripleo-tempest","cherry_pick_of_change":792670,"cherry_pick_of_patch_set":1,"meta_rev_id":"e8ebe1e47a374b8904beae7ef8ebb7d5a4cfd9c7","_number":800648,"virtual_id_number":800648,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 20:39:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-07-19 21:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-07-20 17:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-07-20 17:32:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-14 18:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-19 21:07:20.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-07-20 17:32:58.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7d73976a7f3666ef6a741a56c7d542305cae604a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 15:37:03.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"5fb4d7c19f3b43735b94d78ff543d82f6b662012","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 15:37:21.000000000","message":"Topic tripleo-tempest-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"2feb2b6e0c3a85b53354509b4b4b05b578a60541","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 15:37:24.000000000","message":"Topic set to tripleo-tempest","accounts_in_message":[],"_revision_number":1},{"id":"a4391a58341a8de1ccd0a0704309f9ed7da5a0da","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-07-14 15:37:30.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"58a4cdfc8393a46af5e04f9923db2be603eb1973","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 18:12:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6052474fc21c4604a12f751f19835718 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b155d92f622046ad84e0a2c47004936a : SUCCESS in 4m 06s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/82c77f47398045bbb0540171131a74c5 : SUCCESS in 14m 36s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/52285f54313a4fc8bef959f3e5c01b90 : SUCCESS in 12m 11s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81b8fdb9cbf3451086baa2beaa7010e2 : SUCCESS in 14m 47s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2df2d2bbd41f4779a588766699b6b53a : SUCCESS in 53m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/cba0edd3b65b43f096583660ac4faf7d : SUCCESS in 1h 43m 33s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cccbecf19c154ca2918c665c75fadc2f : SUCCESS in 1h 46m 31s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/0ae2aa321f83456bbb5a555225e5b589 : SUCCESS in 30m 31s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3bd2101adcf4b4e83e1dabb8d67891c : SUCCESS in 34m 26s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cdbf1c7ad2154736a6c5dfd88ac81599 : SUCCESS in 24m 23s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/016ce126eae3417fab73a38099d19adb : SUCCESS in 2h 33m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/e788bd478d1e40f1aa809005f0c19b60 : SUCCESS in 1h 16m 31s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c27ef674c0254fa18d82f464a1df88e0 : SUCCESS in 2h 21m 34s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/729e0d093ac04430a291c97582787f6f : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4c0faa71c9e64ddc5fc61801c04da7c13e602575","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-07-19 21:07:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"707cc9248259fd7c64d4ff99d333f494ecd34905","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-07-20 17:32:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3dbb43463af439f2bf579181e926efc7adf4624e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 17:34:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6657b811a1b6abaa0518d8d631c3cc5e8ae3540f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 20:39:04.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ccefd29a0394403791a7e2a2cf5d99f2 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8c4333ac91c145f7b0ccb77aa13874de : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/e644dc1df4a949ad81663939c75ece21 : SUCCESS in 15m 20s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5cc5d91595a24c1eaab268299b16cf2f : SUCCESS in 11m 54s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5dd68c6c621e4da9981f048567509b48 : SUCCESS in 56m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/f2e45123d4d640eca7bb52441ca6034f : SUCCESS in 1h 27m 54s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/3db67c32cf71452eb6d465e7b6c87550 : SUCCESS in 31m 32s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/29b92f8ff516483291b52d9903530ed6 : SUCCESS in 34m 34s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/231ff4085411472c8497014e6990d21d : SUCCESS in 25m 54s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/42ca24731e924a5d9ec3fcdc9bdee5fa : SUCCESS in 2h 59m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3074df5c90944e708644b32798b189f2 : SUCCESS in 1h 26m 26s\n- tripleo-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/cd2c6667369b44cdaa703e85047895c3 : SUCCESS in 2h 22m 46s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ec8270a046e4141aaf44cd2f7508873 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e8ebe1e47a374b8904beae7ef8ebb7d5a4cfd9c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 20:39:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb93f97fbe39851711f46872d4dcb8be99f7be21","revisions":{"eb93f97fbe39851711f46872d4dcb8be99f7be21":{"kind":"REWORK","_number":1,"created":"2021-07-14 15:37:03.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/48/800648/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/48/800648/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/800648/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/800648/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/48/800648/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/48/800648/1"}}},"commit":{"parents":[{"commit":"1a6af74e3e74f4c8284ccb385554bde0bf136ac2","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a6af74e3e74f4c8284ccb385554bde0bf136ac2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:42:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-07-14 15:37:03.000000000","tz":0},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests in tripleo job, so that we\nrun only subset of test cases used in TripleO repos.\n\nChange-Id: Ia28f458ac84e4f46dcda0dd800276bbd5721a568\n(cherry picked from commit 56c1954629b78921ca9e8347ca7fd77f1d588062)\n(cherry picked from commit 5e0cb3c89b7ae4a78ffcc9c665442dd31a920e35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eb93f97fbe39851711f46872d4dcb8be99f7be21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eb93f97fbe39851711f46872d4dcb8be99f7be21"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
