)]}'
{"id":"openstack%2Fpuppet-swift~803667","triplet_id":"openstack%2Fpuppet-swift~master~I59717abeb5ef120e811511a771f5c3cd491b850f","project":"openstack/puppet-swift","branch":"master","topic":"bug/1939088","hashtags":[],"change_id":"I59717abeb5ef120e811511a771f5c3cd491b850f","subject":"Replace deprecated is_array method","status":"MERGED","created":"2021-08-06 01:50:41.000000000","updated":"2021-08-10 08:14:04.000000000","submitted":"2021-08-10 08:14:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803667-bug/1939088","meta_rev_id":"8f1bd04296ed0ace4208cd80152c90467cda93c4","_number":803667,"virtual_id_number":803667,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-10 08:14:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"all":[{"value":2,"date":"2021-08-10 05:42:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":2,"date":"2021-08-10 06:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2021-08-10 06:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-06 03:23:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-10 05:42:28.000000000","updated_by":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"reviewer":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"state":"REVIEWER"},{"updated":"2021-08-10 06:43:51.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"2607ad7e81686fcef93dbb127c3e80e27bcb7b3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-06 01:50:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"005380a011714c36a90d8608c2bc9e1313155cdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 03:23:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c789b82a5204baaad974f4d8672c4b4 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/25dfae636f9e4c449b67e407aa8fa771 : SUCCESS in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2f1ebef601c422eab19742c9656ee81 : SUCCESS in 4m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5f0e69a633494835bab36c7ca8523c55 : FAILURE in 15m 40s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/84607841a84a4118b94e9634097a2f40 : FAILURE in 14m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d459ebc0537844ea836146f567db1b7c : FAILURE in 12m 19s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3548a02d03d463ca1ddafa0dc9067f1 : FAILURE in 14m 56s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ebd680d880248f2b32d5014b5d640c8 : FAILURE in 12m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7ce3fcc3cac48ecbb7ed565c13a73f8 : FAILURE in 38m 23s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/41136ebed8824c82b5c7cad65525bb24 : SUCCESS in 1h 11m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/db07ca7bbfed4a18baa3a458a7be7ca8 : SUCCESS in 1h 28m 00s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/9e2c6fff2ae24442979714329781e707 : SUCCESS in 26m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e97cb943f38648e9b9c56d6fe66af363 : FAILURE in 24m 17s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/50cf7aec1fb74662bc0e9fbdd31ddc15 : SUCCESS in 25m 44s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba5edc691fb848d88b1573f2e66ff0e0 : FAILURE in 24m 55s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/966c5892962c4ddface5b5ce9f7282ed : FAILURE in 22m 44s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a79536eb4506483484b9f0c99fd7b20c : SUCCESS in 1h 14m 57s","accounts_in_message":[],"_revision_number":1},{"id":"822ee0137b640a03be20fe80e2a2a92522fd288c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-06 03:39:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d5d8f27cc5d5a2d0e612616dca7d5ae8c03b534","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 05:08:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52dbb8f761874956ad8177a81ee3136c : SUCCESS in 4m 11s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52befea596f947f38c1ca0b771b4ca44 : FAILURE in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6cded43b1ec4793a5e663f533d5fca8 : FAILURE in 3m 36s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bae82ff38f304b0abaf6af1b9f0d1527 : FAILURE in 16m 00s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/85e0ed24310e4c9b875de4766980f1c0 : FAILURE in 15m 18s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2cbf94fecbe4dfdaeea504eb066aad9 : FAILURE in 12m 41s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6f6af92b2c24903a046feb204eed55c : FAILURE in 11m 09s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/63f5be0a2a164b26bf3211e06b91be85 : FAILURE in 12m 58s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ce10d02349249b59e9da0560c8fbd97 : FAILURE in 37m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e8e18dc37844877b5e655d231e5f66c : SUCCESS in 1h 25m 00s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9703094a35c2425392097ee279cc82dc : SUCCESS in 1h 06m 44s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f6c6bf6440ad4606919d4a1d5e983ebf : SUCCESS in 30m 54s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/397dacfea0474a519b771f19a2b49ce2 : FAILURE in 22m 39s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d8c4624a24c9440286748668fe360db3 : SUCCESS in 28m 28s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/60813b8369e04e05bf35113108a07cc6 : FAILURE in 21m 07s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fd5f68aeb7447688532d5621dc99060 : FAILURE in 22m 38s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/705455a2141b4d7ea61ac0a020508369 : SUCCESS in 57m 45s","accounts_in_message":[],"_revision_number":2},{"id":"79a0fedad4be0707c65eff7dd6a12404e60e48ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-06 05:13:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"01abfc868e0a51e3d9c88df014037a8b8cef1648","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 06:47:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5799332b655946c7b1bc1ae72bb0df55 : SUCCESS in 4m 37s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebba979bda1b450da8b1532475f284c4 : FAILURE in 4m 26s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/330501c08cdf4f4bb7541c1f9c6a62d3 : FAILURE in 3m 31s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1482a5c70d67415f8c1285b08b591bc6 : FAILURE in 16m 15s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/aab75f94b2ca4711aaddb6a7cef837eb : FAILURE in 16m 24s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0625ebf57a3841258fc2b13fdb88f144 : FAILURE in 12m 54s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/685915f11be54dbeb3d15bbd46811f81 : FAILURE in 16m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3e36b0b2c1e74935bf8b223c88073eca : FAILURE in 12m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15daea47dd9e4829b02abcd0c0416497 : FAILURE in 40m 53s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/2daabd79a91c48f6a9440fdc7f2b2ce5 : SUCCESS in 1h 26m 32s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0d2aa84a663047be80529debc7f8e019 : SUCCESS in 1h 17m 57s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/1ce9cb92e8cc4aa397fbbcb99ddf1159 : SUCCESS in 27m 25s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54a5bc2b1e0347aa91166efc393747aa : FAILURE in 24m 38s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/6775e88365a74a5480734034d1ac0402 : SUCCESS in 27m 37s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5092a51aba5b4ef58be2b9a554e63599 : FAILURE in 26m 11s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d622e618d8184b55ac870b3c572ed1ee : FAILURE in 28m 56s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/eeb89e41551241debff3b7c8765506bd : SUCCESS in 1h 11m 31s","accounts_in_message":[],"_revision_number":3},{"id":"6b0ffcb65a0db5ce0fe5c8dfb97fc27aac6ca7cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-09 23:51:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25ce46508c2a198f2ee5a3b8fd33e4954f296dff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 01:28:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e92568dfceeb4f60829719559ccdda4c : SUCCESS in 3m 40s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33c56b9f2f9c4d3e922740710ea32afd : SUCCESS in 3m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c1c7097ce85a4ec39ba5d7af19f67f91 : SUCCESS in 3m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/d91ca74e9c144fb695273a53376a40a3 : SUCCESS in 15m 15s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/66fa11791a7f4e34808dc50206a23f0c : SUCCESS in 15m 04s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99c76ecce11a49b886b446cb41406786 : SUCCESS in 13m 16s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3bbb277a70a34ccf823a59dc01b174cb : SUCCESS in 12m 48s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f6e49c179cb4cccab0600b462156e43 : SUCCESS in 13m 09s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1b023e72b034cc889ca904f765b64ab : FAILURE in 38m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/1b922b55c10e4618937627db496e0656 : SUCCESS in 1h 30m 26s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/05f5e4c7646f4faab22392ac66b84fad : SUCCESS in 1h 09m 37s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e4bfab070664255a79b4a6870e3f2d2 : SUCCESS in 27m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77d3bf4737c1442eb1c2c8a1fe9212b7 : FAILURE in 25m 50s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/df38e30d42514668951c803bb54a80e1 : SUCCESS in 27m 28s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fc511cd5a3ab422499111260dadc96e7 : FAILURE in 26m 42s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1224f0e61f4341498ed7dbc6d6e3062c : FAILURE in 26m 12s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b0ee262a62a6463489da387b7b2a1876 : SUCCESS in 1h 14m 01s","accounts_in_message":[],"_revision_number":4},{"id":"774f3bb907e34841901fef0083ed90496e09ddad","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2021-08-10 05:42:28.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"bee8c3074f037bc2a0af4b8dfdedcea96cc91e1e","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-08-10 06:43:51.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5eacfc422046a463541b0d0590e6a23409cada35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 06:45:02.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ddc5a6c093f979cf8499f4c4190d25b9258f2870","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:14:04.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/468efefaf00b4908bb13db168c62d7d0 : SUCCESS in 5m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76f57b1f2e9448b38390a0c3b187240a : SUCCESS in 5m 04s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff20244962b048568c4c634f171d2789 : SUCCESS in 5m 39s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b795a8d9369b4abba013c495cb630ce7 : SUCCESS in 17m 11s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/f567c2a3f4c04052b7a68ac49df89a92 : SUCCESS in 16m 39s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f94c79fca0fe43c39c934b59b8fee15d : SUCCESS in 14m 36s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/48023c48cb4d4150a280260b2a482190 : SUCCESS in 15m 35s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4160900a46da4cf3b176f594f6f21e93 : SUCCESS in 1h 25m 07s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/42d2aa4912c74a758cf471de6f848276 : SUCCESS in 30m 34s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1bd1d78443e748eebd41a8a2a0411112 : FAILURE in 26m 11s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/5ed26ecdaad44898acdcbdff6ee2b847 : SUCCESS in 1h 14m 43s","accounts_in_message":[],"_revision_number":4},{"id":"8f1bd04296ed0ace4208cd80152c90467cda93c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 08:14:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7d7d8679d6dc249e85ceb90f82d9017414414c0b","revisions":{"be98af16ddf5208bc391a502dcc3a83c2804ecb6":{"kind":"REWORK","_number":1,"created":"2021-08-06 01:50:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/803667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/803667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/1"}}},"commit":{"parents":[{"commit":"e210b64e6d04ee1ce46f2266be52106c94f80b3e","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e210b64e6d04ee1ce46f2266be52106c94f80b3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:50:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:50:15.000000000","tz":540},"subject":"Replace deprecated is_array method","message":"Replace deprecated is_array method\n\nChange-Id: I59717abeb5ef120e811511a771f5c3cd491b850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/be98af16ddf5208bc391a502dcc3a83c2804ecb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/be98af16ddf5208bc391a502dcc3a83c2804ecb6"}]},"branch":"refs/heads/master"},"643cef4f8ad38f1115c56e4c25e771894b2e3ded":{"kind":"REWORK","_number":2,"created":"2021-08-06 03:39:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/803667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/803667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/2"}}},"commit":{"parents":[{"commit":"e210b64e6d04ee1ce46f2266be52106c94f80b3e","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e210b64e6d04ee1ce46f2266be52106c94f80b3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:50:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 03:39:24.000000000","tz":540},"subject":"Replace deprecated is_array method","message":"Replace deprecated is_array method\n\nCloses-Bug: #1939088\nChange-Id: I59717abeb5ef120e811511a771f5c3cd491b850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/643cef4f8ad38f1115c56e4c25e771894b2e3ded"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/643cef4f8ad38f1115c56e4c25e771894b2e3ded"}]},"branch":"refs/heads/master"},"04ca9b73005bc523a8c9a7adbc8bb32e22d4065f":{"kind":"REWORK","_number":3,"created":"2021-08-06 05:13:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/803667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/803667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/3"}}},"commit":{"parents":[{"commit":"e210b64e6d04ee1ce46f2266be52106c94f80b3e","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e210b64e6d04ee1ce46f2266be52106c94f80b3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:50:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 05:13:51.000000000","tz":540},"subject":"Replace deprecated is_array method","message":"Replace deprecated is_array method\n\nCloses-Bug: #1939088\nChange-Id: I59717abeb5ef120e811511a771f5c3cd491b850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04ca9b73005bc523a8c9a7adbc8bb32e22d4065f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/04ca9b73005bc523a8c9a7adbc8bb32e22d4065f"}]},"branch":"refs/heads/master"},"7d7d8679d6dc249e85ceb90f82d9017414414c0b":{"kind":"REWORK","_number":4,"created":"2021-08-09 23:51:53.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/803667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/803667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/803667/4"}}},"commit":{"parents":[{"commit":"e210b64e6d04ee1ce46f2266be52106c94f80b3e","subject":"Merge \"Drop Fedora support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/e210b64e6d04ee1ce46f2266be52106c94f80b3e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-06 01:50:15.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-09 23:51:43.000000000","tz":540},"subject":"Replace deprecated is_array method","message":"Replace deprecated is_array method\n\nCloses-Bug: #1939088\nChange-Id: I59717abeb5ef120e811511a771f5c3cd491b850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7d7d8679d6dc249e85ceb90f82d9017414414c0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7d7d8679d6dc249e85ceb90f82d9017414414c0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
