)]}'
{"id":"openstack%2Fpuppet-swift~804982","triplet_id":"openstack%2Fpuppet-swift~master~I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81","project":"openstack/puppet-swift","branch":"master","topic":"keystone-roles","hashtags":[],"change_id":"I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81","subject":"Add system/project reader roles parameters","status":"MERGED","created":"2021-08-18 10:33:16.000000000","updated":"2021-09-02 17:49:53.000000000","submitted":"2021-09-02 17:48:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"804982-keystone-roles","meta_rev_id":"fab02f07ac2317b90f8cbf4325455a5fa8e77ba9","_number":804982,"virtual_id_number":804982,"owner":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 17:48:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-08-30 11:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":2,"date":"2021-09-02 16:21:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":1,"date":"2021-09-02 16:21:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-18 12:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-27 13:22:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-09-02 16:21:54.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"4aebabbeb0606d23b2238bec66e7224de6b9cb96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-18 10:33:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"526025293a87d802de6cc24b22af152efac89aaf","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-18 10:36:37.000000000","message":"Patch Set 1: Workflow-1\n\nWIP","accounts_in_message":[],"_revision_number":1},{"id":"81b56e5dc0127684900f9b137c0b07ba5798ca70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:04:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec9b410b1e2742f2b10229947c9d107f : SUCCESS in 3m 43s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d001ce66c434a81acc2f3367eae883a : SUCCESS in 3m 34s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05447fb044a4435d894b98d7e2751a54 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8a8e6e234e24c40902b14d5b08a5573 : FAILURE in 9m 48s\n- puppet-openstack-unit-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/4462fbed79bb41d4b160d58597a481b9 : FAILURE in 9m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f88c3cdb99e44056967a4f1be31895eb : FAILURE in 7m 43s\n- puppet-openstack-unit-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/908f27447c7f470db44ea79cee5b3276 : FAILURE in 6m 43s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaaad15a523a45f8b16500330c95b7af : FAILURE in 7m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fbeb7f14613343d896e1f9564a1cfcfb : FAILURE in 37m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/9016c9be98204d62a1474c5f93a6bc2e : SUCCESS in 1h 23m 01s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b06b899326494001bed45d3a523d713f : SUCCESS in 1h 25m 56s\n- puppet-openstack-litmus-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/116bc9394be24f8ab459dbad6a6762b4 : SUCCESS in 27m 19s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30a859076a4b41f7bec664b78ea1c4df : FAILURE in 23m 50s (non-voting)\n- puppet-openstack-litmus-7.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/0eba68dab1264a5dbe8af327509bb764 : SUCCESS in 28m 06s\n- puppet-openstack-litmus-7.5-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee81a48869c34bd0b3c6424e640fef5b : FAILURE in 25m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f765b6067e14672a58f5d2b028cdc22 : FAILURE in 22m 54s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46c59b5d8fbf4117bba2a3a6feaa5341 : SUCCESS in 3m 42s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a4ecc15db6454d33835b6a31750d4b1b : SUCCESS in 1h 22m 38s","accounts_in_message":[],"_revision_number":1},{"id":"06c488cb6d8d526e97a5416f4ace06f9ac2bbc63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-24 12:31:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"613aba6f2175cee963ee736515b82b4ec937223d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 13:56:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec48cc7e104c4724a09e46498a964e4f : SUCCESS in 4m 16s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f00bbf95cf446568c71223eaa23207c : SUCCESS in 4m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad09f9867d954d92829247ad29163f9b : SUCCESS in 5m 25s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf307a2b0f51482e9b53a3ff588eb4db : SUCCESS in 16m 50s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a7312acaad84fceb07110b5fbfb9f4c : SUCCESS in 18m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a088bc04204a44b598d7c8245cc77ed3 : SUCCESS in 17m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae5c24877724b0a9c677ffba431c00b : SUCCESS in 16m 06s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8704988391c541b19f7b5578bbfb1c86 : SUCCESS in 16m 08s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d66589602bf4bd79073c432493463dc : SUCCESS in 54m 32s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf5fda79b91e4d58a91a17ea5ff3ea88 : SUCCESS in 1h 23m 30s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a05517f66d54bb6ab2eb081b8fc263c : FAILURE in 26m 21s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/75ec0ae1dda649dbba8caaacddd6057e : SUCCESS in 32m 42s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3fce00213044cc59ff15233206949e9 : FAILURE in 28m 02s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfa426b8ca7a4aa696b9d95bd829c7bf : FAILURE in 26m 42s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b76e2b5a52b43d3a58a0916cb17a40a : SUCCESS in 5m 31s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/a3ef20f9a4ed4907a3d77bc174690f80 : SUCCESS in 1h 16m 53s","accounts_in_message":[],"_revision_number":2},{"id":"abcc10c9d295de33ac53c8dea998e70394939e90","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-26 11:32:48.000000000","message":"Topic keystone-roles removed","accounts_in_message":[],"_revision_number":2},{"id":"1f7f61779cccf1604994ef1d17c3c8749a21cd2d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-26 11:32:50.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":2},{"id":"daa7023aa1910761b5198e3a8bc1079219123ffb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-27 13:22:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nThank you for making these changes. Please find my comments inline.","accounts_in_message":[],"_revision_number":2},{"id":"ad76c74c5223042746127654d2d0c3904e1b2b01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-30 09:52:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9c0c358af11772e717d4a828efdf31c2281742f2","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-30 09:53:24.000000000","message":"Patch Set 3:\n\nThanks Takashi for your review! I just updated the patch, please have a look.","accounts_in_message":[],"_revision_number":3},{"id":"b3579b620b1d42a1f9b0c1b39eae40ce3e22719d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 11:18:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8cdfa3b14653432aa5e63142be2ef0ec : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8127a2ecff1f4d0aa3976956ebccab25 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/595e53944c024804876d1bd7ab30fb0d : SUCCESS in 4m 33s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bafec6de07634129b6b250611776b28e : SUCCESS in 18m 01s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc9f2ef5567b4b8eaa182c163bf91be4 : SUCCESS in 22m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/114be8544a604f0eb48076525385910e : SUCCESS in 14m 33s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57850597f58446fb9482989aa75b3953 : SUCCESS in 16m 59s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/966c86b0ba464a40a0493f4c5e1f4455 : SUCCESS in 16m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa48c58ce468495ab41305919d046629 : SUCCESS in 52m 19s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c9df22e85784750a590942dec987fab : SUCCESS in 1h 24m 24s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cde21a559d204746a8643713e4d2c141 : FAILURE in 26m 36s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/088d7f8a9f604b37ad24254c4c422bf1 : SUCCESS in 29m 45s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b851f81518d84ff5a5cb124bda6acc6a : FAILURE in 29m 23s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66d7707af7174f6f9dab39cbb3371afb : FAILURE in 26m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/362d23250ef242b49383e2ce689ed14e : SUCCESS in 5m 02s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/b9b7c17df73244e999baf07d78d96027 : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":3},{"id":"b87454275c16356ee22d40b51693e510206a0c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 11:25:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"1ca23606ec624ff04649c4d78d08e3b3a79274f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 11:26:51.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"844f78dde26c9f90c0a6e1fcac57b998ecec5f5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-30 11:28:01.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks Christian.\nI have made slight changes to address the following points, but let me know if you have any concern.\n - make parameter orders consistent in all places \n - validate the default value of the new parameters","accounts_in_message":[],"_revision_number":5},{"id":"ba3813193dd1c15bfafc863fb33c5e94d5f5daf7","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2021-08-30 12:35:17.000000000","message":"Patch Set 5:\n\nThanks Takashi, LGTM!","accounts_in_message":[],"_revision_number":5},{"id":"44f7378f2c0cc922dce72f0d3acf1519070efb86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 12:58:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d8df5f59dc2a483993bed989c71f23c6 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c11ec5e3a84447cf86f4e11646574503 : SUCCESS in 5m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d946876e38654281a9a746393303b149 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/0323e309114d455f9834a3944a40dac0 : SUCCESS in 16m 09s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/239629d80d714f93bbcbca1dc2a02695 : SUCCESS in 15m 31s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de3731e2b99497386075a52a2f6370c : SUCCESS in 14m 34s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52fbff7522494072b4b57a53a6fc51a8 : SUCCESS in 13m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16268233749a483a90dc9d9514c7bc6f : SUCCESS in 14m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b95cf6dd24c045f38cadedb3911d171f : SUCCESS in 53m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a663ba60618b44f4b0a0a6e776ae39e6 : SUCCESS in 1h 25m 21s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b19ff18ecd94c71ba58f1e76647017d : FAILURE in 25m 00s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4f0c9f12659f4b3ea3a921b3eb2545ef : SUCCESS in 27m 42s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d786f16391ce4b4eb7bee2be7b5907ea : FAILURE in 24m 05s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fc3b1358dcc4cd2a688af36307e3abc : FAILURE in 26m 51s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/30cfaf8324654b6e8662adca282b9bb4 : SUCCESS in 3m 52s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/98a98e3c35d947e1b06c561bc912aef9 : SUCCESS in 1h 22m 29s","accounts_in_message":[],"_revision_number":5},{"id":"c095806722ee6e3828ad11f26bd5004f16abe6c8","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-02 16:21:54.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f0788bc7f68d9d4b751cd91583350cfc0fdf75f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 16:23:07.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8c2e4b844801305630bfae4e14ca3e6d73725c4c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:48:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1d162d496d4b4a379d705384acd4afa3 : SUCCESS in 6m 08s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7d6be0d63ea4ac4ba5c07e2f330e327 : SUCCESS in 5m 43s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1424961274a44c3dae30d3671db98b93 : SUCCESS in 5m 11s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0f0c9458c5f4e1ca921c9e27306dc60 : SUCCESS in 19m 04s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fbbe8d469ba4405c93a1fe43e997ed1f : SUCCESS in 16m 01s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0218e397ae64d22956e4d4d7463127a : SUCCESS in 15m 44s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b784b474eff743ba8bfc60db22d97199 : SUCCESS in 14m 31s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3232e739d93d46a1a56378d2acf4e3a1 : SUCCESS in 1h 23m 35s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3178b38adc994ea18520286456e71de2 : SUCCESS in 35m 09s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1341fc7ff67748e98c8e57b4b287a9cf : FAILURE in 24m 31s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8803c920551744d1aae94199dc9b21f1 : SUCCESS in 4m 15s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8c93575151604068a2caef5d7723c18f : SUCCESS in 1h 21m 24s","accounts_in_message":[],"_revision_number":5},{"id":"c36b0c2a04967810c1f83381aa2c0976c6522969","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:48:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"fab02f07ac2317b90f8cbf4325455a5fa8e77ba9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:49:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fed2cf9d08094baeaa6c9cd781bd4128 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"969babc71e519442a59c1535a4afec4200e52676","revisions":{"31c3a6534f939918495e8a208f1b4ed8215c3de5":{"kind":"REWORK","_number":1,"created":"2021-08-18 10:33:16.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/804982/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/82/804982/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/1"}}},"commit":{"parents":[{"commit":"787a87d24fee180f607d965182258072747be65e","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/787a87d24fee180f607d965182258072747be65e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-18 10:32:59.000000000","tz":120},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/31c3a6534f939918495e8a208f1b4ed8215c3de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/31c3a6534f939918495e8a208f1b4ed8215c3de5"}]},"branch":"refs/heads/master"},"382326405f2b423f56d353a2ce019321cbb4f2d2":{"kind":"REWORK","_number":2,"created":"2021-08-24 12:31:23.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/804982/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/82/804982/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/2"}}},"commit":{"parents":[{"commit":"787a87d24fee180f607d965182258072747be65e","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/787a87d24fee180f607d965182258072747be65e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-24 12:29:10.000000000","tz":120},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/382326405f2b423f56d353a2ce019321cbb4f2d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/382326405f2b423f56d353a2ce019321cbb4f2d2"}]},"branch":"refs/heads/master"},"ce55aa6b2a0247e995ba5bd2e6f440eea28ef5d4":{"kind":"REWORK","_number":3,"created":"2021-08-30 09:52:25.000000000","uploader":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"ref":"refs/changes/82/804982/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/82/804982/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/3"}}},"commit":{"parents":[{"commit":"787a87d24fee180f607d965182258072747be65e","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/787a87d24fee180f607d965182258072747be65e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-30 09:24:50.000000000","tz":120},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ce55aa6b2a0247e995ba5bd2e6f440eea28ef5d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ce55aa6b2a0247e995ba5bd2e6f440eea28ef5d4"}]},"branch":"refs/heads/master"},"487110ee0b954b9929615288cf2877dea41ad972":{"kind":"REWORK","_number":4,"created":"2021-08-30 11:25:59.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/804982/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/82/804982/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/4"}}},"commit":{"parents":[{"commit":"787a87d24fee180f607d965182258072747be65e","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/787a87d24fee180f607d965182258072747be65e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-30 11:25:49.000000000","tz":540},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/487110ee0b954b9929615288cf2877dea41ad972"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/487110ee0b954b9929615288cf2877dea41ad972"}]},"branch":"refs/heads/master"},"969babc71e519442a59c1535a4afec4200e52676":{"kind":"REWORK","_number":5,"created":"2021-08-30 11:26:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/82/804982/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/82/804982/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/82/804982/5"}}},"commit":{"parents":[{"commit":"787a87d24fee180f607d965182258072747be65e","subject":"Merge \"Support Puppet 7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/787a87d24fee180f607d965182258072747be65e"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-30 11:26:43.000000000","tz":540},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/969babc71e519442a59c1535a4afec4200e52676"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/969babc71e519442a59c1535a4afec4200e52676"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
