)]}'
{"id":"openstack%2Fpuppet-swift~805522","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~Ia28f458ac84e4f46dcda0dd800276bbd5721a568","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"tripleo-tempest","hashtags":["puppet-tripleo-job"],"change_id":"Ia28f458ac84e4f46dcda0dd800276bbd5721a568","subject":"TripleO: Reduce tempest test scope","status":"MERGED","created":"2021-08-22 05:15:57.000000000","updated":"2021-08-22 10:21:46.000000000","submitted":"2021-08-22 10:21:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805522-tripleo-tempest","cherry_pick_of_change":800648,"cherry_pick_of_patch_set":1,"meta_rev_id":"6b4b20e540e14f62f4528d60f79efc9bb8fe7c8e","_number":805522,"virtual_id_number":805522,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-22 10:21:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-08-22 05:21:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-08-22 05:21:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-22 07:53:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eedc8e7e598134591411e5e769fe1875d24bb5db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:15:57.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"647d09564a412be08fd8dfdf17067f72a7a473b9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:16:02.000000000","message":"Topic tripleo-tempest-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"e327baf50269b7830397cd1155309c7821077272","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:16:04.000000000","message":"Topic set to tripleo-tempest","accounts_in_message":[],"_revision_number":1},{"id":"1eca687d7e6fcbff7ac8871b9f6356fcd05b4adc","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:16:09.000000000","message":"Hashtag added: puppet-tripleo-job","accounts_in_message":[],"_revision_number":1},{"id":"8b72cce56a26c13926fd796f031eed8521642c37","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-08-22 05:21:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1485c6d71e384a278a3389ddfd964409f4e68a42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 07:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b2dd39624ddc4ebbb0c272aff92e75ef : SUCCESS in 4m 04s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/684053d1b1bc49abad862bac96ccaad7 : SUCCESS in 3m 27s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/caada4c7a31d4ea2bdaaf5474e12a971 : SUCCESS in 3m 27s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/efd05feaee2948c38f95a5f772e4f596 : SUCCESS in 16m 27s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/e7b8bdb184bc4cacb4c19f4f1bb5ba2c : SUCCESS in 15m 54s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d9c10827dd3748109b6c171fa1ac9384 : SUCCESS in 11m 17s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5f12df77ebd94f77aa24e03ef5167dce : SUCCESS in 10m 41s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/24e453fe9f63447092948f3a56173322 : SUCCESS in 49m 00s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/1ccc3500792a416e87b29a11f7d13c1e : SUCCESS in 58m 53s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/4baa3030e1c34cbc9f90a739aa4ffe19 : FAILURE in 40m 21s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/e4e40aab69224ebeb6d5282b527ef9bf : SUCCESS in 29m 20s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/205364aaf4a949aa9906e139df0dbf52 : SUCCESS in 29m 16s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/354012b2994241418325c49c55ea7668 : SUCCESS in 21m 30s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/b989fc582d914ed1b8c14968009436d5 : SUCCESS in 2h 32m 59s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5e5b6c0f31614b0db5946157e0ffd5ad : SUCCESS in 1h 19m 40s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/603b12849b4b4174b6c83f236ac6babc : SUCCESS in 1h 58m 46s","accounts_in_message":[],"_revision_number":1},{"id":"2d653196ac944530adea88b2b71d0f4a0a893a1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 07:54:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75f19cfd8a331ce27237b5d5096510182b6f8294","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 10:21:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa3e4c3d5919477e9034e1644a599bae : SUCCESS in 3m 28s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c7140cc1969441a3b07376d875018cb1 : SUCCESS in 3m 45s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9a594c5828fe45c8b877469de26a0797 : SUCCESS in 3m 23s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/cb3d5693f0fe4b018784e73c63dd67ee : SUCCESS in 15m 30s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/8aef8dfb88ec4612ae4872e61a125e4e : SUCCESS in 15m 34s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a63c71a1e5144c9bb77008e51558d92 : SUCCESS in 9m 53s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83a0625d3edc4d5fa2420eae329c3dad : SUCCESS in 49m 32s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/a3aad2e87e154fbabbd0cf5077f15feb : SUCCESS in 1h 03m 40s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/18eb044ee60f4875a228ad8901832747 : SUCCESS in 29m 04s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/5f809d5ae57a468cbbdd89737873a67d : SUCCESS in 2h 25m 16s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/eb34773ce0434d0496f3bfd75c0a6836 : SUCCESS in 1h 12m 06s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/44c27f87619944939a6e6926c03fae46 : SUCCESS in 2h 10m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6b4b20e540e14f62f4528d60f79efc9bb8fe7c8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-22 10:21:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"efee875a95dcb12626507e7c3395407d002f84b1","revisions":{"efee875a95dcb12626507e7c3395407d002f84b1":{"kind":"REWORK","_number":1,"created":"2021-08-22 05:15:57.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/22/805522/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/22/805522/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/805522/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/805522/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/22/805522/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/22/805522/1"}}},"commit":{"parents":[{"commit":"2b267de322a09b234d60c4e56ff82262779219b5","subject":"Set fixture paths for unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/2b267de322a09b234d60c4e56ff82262779219b5"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-05-14 02:42:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-22 05:15:57.000000000","tz":0},"subject":"TripleO: Reduce tempest test scope","message":"TripleO: Reduce tempest test scope\n\nThis change reduces scope of tempest tests in tripleo job, so that we\nrun only subset of test cases used in TripleO repos.\n\nChange-Id: Ia28f458ac84e4f46dcda0dd800276bbd5721a568\n(cherry picked from commit 56c1954629b78921ca9e8347ca7fd77f1d588062)\n(cherry picked from commit 5e0cb3c89b7ae4a78ffcc9c665442dd31a920e35)\n(cherry picked from commit eb93f97fbe39851711f46872d4dcb8be99f7be21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efee875a95dcb12626507e7c3395407d002f84b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efee875a95dcb12626507e7c3395407d002f84b1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
