)]}'
{"id":"openstack%2Fpuppet-swift~807069","triplet_id":"openstack%2Fpuppet-swift~stable%2Fwallaby~I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81","project":"openstack/puppet-swift","branch":"stable/wallaby","topic":"keystone-roles","hashtags":[],"change_id":"I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81","subject":"Add system/project reader roles parameters","status":"MERGED","created":"2021-09-02 23:37:45.000000000","updated":"2021-09-09 18:13:09.000000000","submitted":"2021-09-09 18:11:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807069-keystone-roles","cherry_pick_of_change":804982,"cherry_pick_of_patch_set":5,"meta_rev_id":"b031de4377dac31e28a8af12b0e130e1f8fbc92d","_number":807069,"virtual_id_number":807069,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 18:11:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-02 23:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-09-09 16:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-09-09 16:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 23:37:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2021-09-03 01:13:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 16:17:08.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"5e710da4850b7a64a6de7d650451f1060215284a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:37:45.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"407c18d1f7bc8801705c1433fad2f7e91c66953c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:37:50.000000000","message":"Topic keystone-roles-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"fd4d424d7b9f41b92c67b51c09634203e1110aff","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:37:55.000000000","message":"Topic set to keystone-roles","accounts_in_message":[],"_revision_number":1},{"id":"f3b06ecbd71d1f9e58c19d51aa779731cb123383","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-02 23:41:43.000000000","message":"Patch Set 1: Code-Review+2\n\nThe parameter is implemented during Wallaby and this is a clean backport from master.\n\nI\u0027d leave my +2 for fast approval considering the low risk of backporting this to Wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"a229ca7308647e7a9d2c48f2b46ff8cd319a1467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 01:13:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e10a81f0bf047f09868f458c44af881 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/309915234c8a43e8acd315047c0057a7 : SUCCESS in 4m 03s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/09f7c4b7d43e479a97ee77a5e39dce2a : SUCCESS in 16m 17s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb9e67ff75354518a0b19f5f46bbb9bd : SUCCESS in 14m 29s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e842bba0d3114992aa1fa399adca0589 : SUCCESS in 13m 26s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/257d75a70f344a6bb1b1a3dd3882dad7 : SUCCESS in 53m 25s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e398a71e69048b3bbd6cad86e2f713d : SUCCESS in 1h 29m 12s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/5d78e4fb8d064ca5b66b274176015002 : SUCCESS in 29m 20s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c00c0611a4fb431c8c65b42ddf533885 : FAILURE in 24m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5fcae2a24af45629f17306a102a45cf : SUCCESS in 4m 16s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/945f3085ba87442b96f542880c53a3a8 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7eeb628866587734ca0f767fef85a63958223e51","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-09 16:17:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6073a34c6eab528fc9835250c182f51d61913c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:18:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b49f483a16a21829f88ff5282ab7aad598ae2f00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 18:11:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7c1b47aaacea4b21924dab0bf4dd1b0a : SUCCESS in 5m 06s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bec2cb41240c4d46b7a7c8f2ed84c08d : SUCCESS in 4m 37s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8d92628d23684b62ba2db65664d6d9c0 : SUCCESS in 18m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20832395630342fab9e8f83af716b9d4 : SUCCESS in 17m 17s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bb239e9b09994240b40dac9994511dab : SUCCESS in 1h 33m 48s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/03b1b53f882747b48291df3a47225829 : SUCCESS in 30m 28s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7a4bea181b23458eb1aefba0be15df0f : FAILURE in 27m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1b4a5da7beb840438c522ba91a64e609 : SUCCESS in 5m 20s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ef78b9dbb50e4faab7afb9dcab490a7b : SUCCESS in 1h 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"cb831f7d8fe346e7f9aff366c4080257abbb6da8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 18:11:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b031de4377dac31e28a8af12b0e130e1f8fbc92d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 18:13:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/300bb013d4d54aa2b6acf1b21c21b53b : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"236a48239140a028ed12fcdb641734f960d3f5b2","revisions":{"236a48239140a028ed12fcdb641734f960d3f5b2":{"kind":"REWORK","_number":1,"created":"2021-09-02 23:37:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/807069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/69/807069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/807069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/807069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/807069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/69/807069/1"}}},"commit":{"parents":[{"commit":"c4c80d1f183d1a1ece9dc935b6cc651b77421bc2","subject":"Replace deprecated is_array method","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/c4c80d1f183d1a1ece9dc935b6cc651b77421bc2"}]}],"author":{"name":"Christian Schwede","email":"cschwede@redhat.com","date":"2021-08-17 09:31:01.000000000","tz":120},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-02 23:37:45.000000000","tz":0},"subject":"Add system/project reader roles parameters","message":"Add system/project reader roles parameters\n\nChange-Id: I34640f6245ad94d32c0a8eba46bf5ffc5e2efb81\n(cherry picked from commit 969babc71e519442a59c1535a4afec4200e52676)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/236a48239140a028ed12fcdb641734f960d3f5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/236a48239140a028ed12fcdb641734f960d3f5b2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
