)]}'
{"id":"openstack%2Fpuppet-swift~807412","triplet_id":"openstack%2Fpuppet-swift~stable%2Fvictoria~I9ef8fd94fbbbb909c992cf3b16cd1b5670b10845","project":"openstack/puppet-swift","branch":"stable/victoria","hashtags":[],"change_id":"I9ef8fd94fbbbb909c992cf3b16cd1b5670b10845","subject":"Revert \"Add content provider jobs for tripleo based jobs\"","status":"MERGED","created":"2021-09-04 03:28:21.000000000","updated":"2021-09-05 17:39:20.000000000","submitted":"2021-09-05 17:39:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807412","meta_rev_id":"098e60c319acaf87dc97a2b463d504630ff34f41","_number":807412,"virtual_id_number":807412,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-05 17:39:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-04 13:24:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-09-04 13:24:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 06:21:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a57c211b8bccb2a9da1ab68a31a46f76945dea00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 03:28:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ff6f0052e00aff96f81ce17a21e1ae86021b1ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 03:29:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"fb0bf8891346bd3638f96e992efe1aca3b2d3fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 06:21:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71c4fc2f4d0f42a1a86c1c059fd00124 : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e31403cd4ba415e9e2fb1506c6cb847 : SUCCESS in 4m 07s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/3ebde114d9c24ba6b63d34a951a922bc : SUCCESS in 14m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f089ddc7bcd64ea09a98c5ff219e6957 : SUCCESS in 11m 41s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a31ef22b6704c9b9b5b1aeb43eba15d : SUCCESS in 13m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fc59a7c4da7e4d999bab6db177adf403 : SUCCESS in 55m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/00f1e68d5b3b40e8839117abf8db660f : SUCCESS in 1h 46m 37s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ac5f9b6148d242369e81571fa6fb9903 : SUCCESS in 1h 30m 30s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a6b0c9eec4be4f479aec4c3987b4e27a : SUCCESS in 29m 50s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/a4ad4862d797493494af9e1f6b76dfdb : SUCCESS in 29m 40s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2f487276d194e35b6120b1c8298899e : SUCCESS in 23m 23s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/48d231fd7876415db11851ad30913abf : SUCCESS in 2h 47m 30s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/d99ec772c23e41b4a5de4d1286c5a0fc : SUCCESS in 1h 24m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e56fed3b52984237a8d7723abc4c6682 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b032b61b959a7768cdf454459fad135e0066515b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 13:24:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3fb4add427e94c2e6b2f17a967cf23e11e87ab0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 13:25:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9fadcef22bce60b7d61d34e8739073f69992e33a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 16:25:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8dbc34a4ee7f42a184e23684d9d85691 : SUCCESS in 6m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d0ae856d8c9c4fe0a400da4d1103fa05 : SUCCESS in 4m 51s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/271f08bfcb16433ab23944578de626f5 : SUCCESS in 18m 45s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b7608aca912491a91e44256f433506f : SUCCESS in 11m 15s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/528a93ccecf24fd08a43c6e8da043e4f : SUCCESS in 51m 45s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/511e1a41b19142e2beb03b37be95b62a : FAILURE in 30m 12s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/4923a4fbe5884c98afc846db041ba252 : SUCCESS in 28m 21s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/b67dc540ae9942c6b989c8c73906683a : SUCCESS in 29m 59s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0a0fed8dd5e34dac88a3f61ab2004fe9 : SUCCESS in 26m 05s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/c563ca3a083249c6adcd6f63762c05ab : SUCCESS in 2h 54m 25s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/33b88ea59b094e0e9f5bd51ad432e2b6 : SUCCESS in 1h 34m 24s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/6ac6cd2e8e674c538d5220488b3977d8 : SUCCESS in 1h 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"95638cb5af64e6fcfb826ef2036dac69062a138b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-05 02:07:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b2b8888a0cbb41f0516d0bfc2b96cd0fffbc06e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 04:52:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c3d0b7628482424aa8389461340247d8 : SUCCESS in 4m 00s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7d5f268797942ecaa73df5155ed18d2 : SUCCESS in 4m 00s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a775272b42141e9a900b3ce9b483c5f : SUCCESS in 16m 10s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e0b4cb34c714ac08854e4778d9db2d8 : SUCCESS in 11m 46s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e1f26cdfb009473b9b2210d92258ea3e : SUCCESS in 14m 14s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/380551bed8524396a5ce1b0226f20759 : SUCCESS in 53m 04s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/73bd9627f74b4a3692397d5f990fb7f1 : SUCCESS in 1h 19m 51s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a52bd333463a4b0d9cfe8be2321e4cf7 : SUCCESS in 1h 10m 53s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c7a5bfff5eaf4ff3928ebf0d4f60f0a7 : SUCCESS in 29m 40s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/54634912f35742df9c3e47c462b38eb2 : SUCCESS in 28m 12s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9cef239698b3423f98a521d37cfe07d6 : SUCCESS in 25m 32s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/483aeda611164ee19ef59da7cd7a1fa1 : SUCCESS in 2h 38m 47s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3f5071ec3d3146a2a92bb74e07053b65 : SUCCESS in 1h 04m 18s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/477408ecb87f47ca9289edcc80d6d5a6 : FAILURE in 21m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4566356d09217f92e09c4f4a9cb9fcd417c164bd","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-05 11:44:19.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c81227d482ee0dd5033204974a78bb8308c10ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 14:44:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/782d9e006eb745009732cc029b2287af : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2955b3da61d74653af3324b71bd0c82a : SUCCESS in 4m 16s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/bc19312484b9426db54665b5e89a624d : SUCCESS in 12m 27s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1850e88e0284acbb4057992e4bbec60 : SUCCESS in 10m 09s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2fa9db4a851b44bf82ef04fa94db8ed9 : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1dd964c362640408b29bd980b827a3e : SUCCESS in 50m 28s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/ecd2194470404247949b90fc72db6406 : SUCCESS in 1h 16m 26s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0e32b112d9084bb081dea488cbecd1fa : SUCCESS in 1h 21m 53s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/c10a89ebd02c4804bb3d475e1d568fa0 : SUCCESS in 26m 58s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/dff4553a484a4d6993c48e912e6eacca : SUCCESS in 29m 59s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ed32bd09e61a4b8cb3a6f1e42b312ea0 : SUCCESS in 25m 30s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/f9219ef254494ad486bfeeb8f0e87c88 : SUCCESS in 2h 55m 46s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/49708f168184455f9536a94bd77d25df : SUCCESS in 1h 36m 33s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/10f5499acb4f44149140b678c9df01de : SUCCESS in 1h 46m 57s","accounts_in_message":[],"_revision_number":2},{"id":"735787c49f478f9064a8aef0393646a0d5832c5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 14:45:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93b112946c1190175252ce41a650cd60ef5d2f99","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 17:39:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3eab3db357c49adabe253a57ad1cfa9 : SUCCESS in 3m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9f68bf32e0943d89b43740c906279db : SUCCESS in 3m 56s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/b2e03c7e27d5484a8398073077dcb131 : SUCCESS in 13m 30s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c054cb7ea7bd4b6188a5f60a9d57d74d : SUCCESS in 12m 49s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c2c47f01af954ca89a1a5e19017428cd : SUCCESS in 53m 05s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/03661820683b4633963b23cb4f736d7f : SUCCESS in 1h 13m 57s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/7b9e63c6f1cc4e23a9997ccbe7b73dc1 : SUCCESS in 29m 03s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/30a9267f9d5448deab24c9ed430336f0 : SUCCESS in 30m 38s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2625a53df8524b19bc27b22bc684d235 : SUCCESS in 24m 34s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/03bf4b949fb640cd99f899d2340b613f : SUCCESS in 2h 49m 40s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/655b9c7a0e0041d98a76732d9d3b9fda : SUCCESS in 1h 19m 15s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8d66feb69c1240cda274893a032796b0 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":2},{"id":"098e60c319acaf87dc97a2b463d504630ff34f41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 17:39:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6922a5e9c44abb3763369721afbf94c0fdc6d3bf","revisions":{"d13df6941f85ceef427f6af76a60bbf75a540cb4":{"kind":"REWORK","_number":1,"created":"2021-09-04 03:28:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/807412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/12/807412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/1"}}},"commit":{"parents":[{"commit":"eb93f97fbe39851711f46872d4dcb8be99f7be21","subject":"TripleO: Reduce tempest test scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eb93f97fbe39851711f46872d4dcb8be99f7be21"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 03:26:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 03:28:08.000000000","tz":540},"subject":"Revert \"Add content provider jobs for tripleo based jobs\"","message":"Revert \"Add content provider jobs for tripleo based jobs\"\n\nThis reverts commit 2a0f3db4be7412b3cdf22e63a44088b2367fd819.\n\nReason for Revert:\nThe change imports the job directly from TripleO repos but we should\nuse the one overridden for Puppet OpenStack repos.\n\nNote for Revert:\nThis change was fixed in stable/wallaby when removing Undercloud job\nso is submitted to stable/victoria directly.\n\nChange-Id: I9ef8fd94fbbbb909c992cf3b16cd1b5670b10845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d13df6941f85ceef427f6af76a60bbf75a540cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d13df6941f85ceef427f6af76a60bbf75a540cb4"}]},"branch":"refs/heads/stable/victoria"},"6922a5e9c44abb3763369721afbf94c0fdc6d3bf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-04 03:29:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/12/807412/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/12/807412/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/12/807412/2"}}},"commit":{"parents":[{"commit":"eb93f97fbe39851711f46872d4dcb8be99f7be21","subject":"TripleO: Reduce tempest test scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/eb93f97fbe39851711f46872d4dcb8be99f7be21"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 03:26:50.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 03:28:35.000000000","tz":540},"subject":"Revert \"Add content provider jobs for tripleo based jobs\"","message":"Revert \"Add content provider jobs for tripleo based jobs\"\n\nThis reverts commit 2a0f3db4be7412b3cdf22e63a44088b2367fd819.\n\nReason for Revert:\nThe change imports the job directly from TripleO repos but we should\nuse the one overridden for Puppet OpenStack repos.\n\nNote for Revert:\nThis issue was fixed during the wallaby cucle when removing Undercloud\njob so is submitted to stable/victoria directly.\n\nChange-Id: I9ef8fd94fbbbb909c992cf3b16cd1b5670b10845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6922a5e9c44abb3763369721afbf94c0fdc6d3bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/6922a5e9c44abb3763369721afbf94c0fdc6d3bf"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
