)]}'
{"id":"openstack%2Fpuppet-swift~807455","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~Iebedb2b7bb702c82946b9436f98c1ae08d4b6b35","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"tripleo-image-build-stable/ussuri","hashtags":[],"change_id":"Iebedb2b7bb702c82946b9436f98c1ae08d4b6b35","subject":"Victoria and before: Build containers only once in TripleO jobs","status":"MERGED","created":"2021-09-04 13:27:54.000000000","updated":"2021-09-09 09:18:27.000000000","submitted":"2021-09-09 09:18:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807455-tripleo-image-build-stable/ussuri","meta_rev_id":"ebf9b44fde2ea51120162a7d106f09d1b6ff6025","_number":807455,"virtual_id_number":807455,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-09 09:18:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-09-09 06:28:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":1,"date":"2021-09-09 06:28:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-04 16:19:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dc51cff273d49daa03df7383a44db1f781588293","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-04 13:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e349e47115ad7ebce78d40a9f889f7959f01972b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-04 16:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/96fa725e34604a28a5acee4ac608119b : SUCCESS in 3m 35s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/84eca08384494b85a8235f27a0026ac5 : SUCCESS in 3m 29s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/655600ad701d406d9813fca32a4282a9 : SUCCESS in 3m 27s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/92ebc083d83247e8ab93339ba6fd2194 : SUCCESS in 16m 06s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/579779a913f6406392cbbae9e02456f8 : SUCCESS in 16m 18s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fdfdfa7ed8b245cfbc36f5f9528842f6 : SUCCESS in 10m 32s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4fd798a294e44793affa18d40c640a5e : SUCCESS in 13m 05s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b5a3780c8d6047aaab05006f2af2fcc8 : SUCCESS in 48m 12s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/2bf662d912a948609c8ba5f66f7325fd : SUCCESS in 1h 10m 25s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/5c4ebbd490034f75a3277587bc92ec56 : SUCCESS in 1h 04m 14s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/b6eb9765f23146aca9043940f0c7d193 : SUCCESS in 32m 26s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/4c4fe6c1393142eb87feb43dd14d4347 : SUCCESS in 29m 29s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1afff0d393a740eba1f5adc3061b04ea : SUCCESS in 22m 23s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/122c6f284bbf4a8bb8333d347b0492bb : SUCCESS in 2h 41m 58s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/1578a9f9bf004addafeba5c5cdae75dd : SUCCESS in 1h 30m 23s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c7c47d0d80144566b647a3324272ecbb : SUCCESS in 1h 01m 46s","accounts_in_message":[],"_revision_number":1},{"id":"decadb9d5530ec4c591ac50e47f684b8cc41146c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-09 06:28:22.000000000","message":"Topic set to tripleo-image-build-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"b9301ce96b51caf2cbac5ddffd605c00f0039698","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-09 06:28:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4cd269d7ee0054ac63ea4f2e98f75ccbbd4cdfe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 06:29:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1439f8fd5eb79ed33e07a07c82b396cf0b769d0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 09:18:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d3938a26f564345abe89974a3ce3f69 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f4a57939d2594da9b1099bec4a4e247a : SUCCESS in 3m 53s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4b4bc0fdac554e97a700704096055fb4 : SUCCESS in 4m 58s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4dc346f9d574f5b8e7953e33de9b5e1 : SUCCESS in 13m 50s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/b0245b085af44fb18d8b38c8c7d7f05d : SUCCESS in 13m 53s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0abe5a4f2afe41b6bc86c5ee20da350a : SUCCESS in 11m 39s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d20ee830025b4d9e98a6017e06eacb19 : SUCCESS in 52m 43s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/5b74e67db9f940eaa3e1e7b383dfbcfd : SUCCESS in 1h 11m 58s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/27a21221d2b743a68c618785d36655d9 : SUCCESS in 33m 31s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/db5c61f78bc240929c7b8a46f5f44665 : SUCCESS in 2h 43m 22s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/ef18662dfbf74f068a1d316b3e467afa : SUCCESS in 1h 25m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/93ae98a6fe2c48359a14fe87650993d4 : SUCCESS in 1h 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ebf9b44fde2ea51120162a7d106f09d1b6ff6025","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 09:18:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ddb37d9e41f850056830a0d4f562f61c00547eca","revisions":{"ddb37d9e41f850056830a0d4f562f61c00547eca":{"kind":"REWORK","_number":1,"created":"2021-09-04 13:27:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/55/807455/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/55/807455/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/807455/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/807455/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/55/807455/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/55/807455/1"}}},"commit":{"parents":[{"commit":"efee875a95dcb12626507e7c3395407d002f84b1","subject":"TripleO: Reduce tempest test scope","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/efee875a95dcb12626507e7c3395407d002f84b1"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-08-22 08:39:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-04 13:27:42.000000000","tz":540},"subject":"Victoria and before: Build containers only once in TripleO jobs","message":"Victoria and before: Build containers only once in TripleO jobs\n\nCurrently container images are built in two TripleO jobs, (standalone\nand undercloud) but this is redundant and inefficient. This change\nmakes container images built by the base content provide job used for\nthese two jobs.\nBecause undercloud job was removed in stable/wallaby this change is\nsubmitted to stable/victoria and older branches.\n\nConflicts:\n\t.zuul.yaml\n\nChange-Id: Iebedb2b7bb702c82946b9436f98c1ae08d4b6b35\n(cherry picked from commit f74af484627e55ba10a333af27b296c045cbfa5e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ddb37d9e41f850056830a0d4f562f61c00547eca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ddb37d9e41f850056830a0d4f562f61c00547eca"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
