)]}'
{"id":"openstack%2Fpuppet-swift~810433","triplet_id":"openstack%2Fpuppet-swift~master~I75a75459b10d0dbf13f72694bc4a70eb4a958e49","project":"openstack/puppet-swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-02-12 16:53:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I75a75459b10d0dbf13f72694bc4a70eb4a958e49","subject":"Add support for [filter:slo] allow_async_delete","status":"ABANDONED","created":"2021-09-22 13:27:46.000000000","updated":"2022-02-12 16:53:34.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb6a652a99157b638baa907435187fa68bb78a65","_number":810433,"virtual_id_number":810433,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-10-04 09:50:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":1,"date":"2021-11-04 11:53:39.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":2,"date":"2021-10-04 13:43:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-22 15:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 17:38:22.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-11-04 11:53:39.000000000","updated_by":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"REVIEWER"}],"messages":[{"id":"615dc7b65b29606d007072981b4aac7d94fd14cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-09-22 13:27:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6668efc14f1b2d44033c475c7d546b49cadee355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 15:31:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0fc4d6a959dc4215a2e977df3b7bcb23 : FAILURE in 5m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4700ff15daf542159d6f0f059ecb635e : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fb92eb5df91426f8c16fccbdcf2a0f8 : SUCCESS in 5m 02s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3f016c445564801ab6906d14c62bbc9 : SUCCESS in 16m 43s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7e66afc001e6466aa4a830f66304ead8 : SUCCESS in 16m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e895948481b494e999e64cfb10cd853 : SUCCESS in 13m 31s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3c794c8a74e40018d0222706d81b692 : SUCCESS in 15m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bbb8553f217242308be14ec1273aa448 : SUCCESS in 13m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc128bccf8144705af780419c5af0303 : SUCCESS in 52m 22s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/eed39f548f264aa6a549bf7d1605d2b0 : SUCCESS in 1h 32m 23s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e075748505442a89ac351f9c63d259c : FAILURE in 25m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/0b2225e9feb946f38b18d5ff403b94ed : SUCCESS in 31m 22s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e74192743ed4cd59a24d78794a002f6 : FAILURE in 30m 50s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0757eb5a5b324fc690e1dd6c41f566a9 : FAILURE in 23m 48s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36bb6b2e83ba448b9e0a868920e10a79 : SUCCESS in 5m 25s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7aaea240395d402b838988954f8a4dc5 : SUCCESS in 1h 19m 06s","accounts_in_message":[],"_revision_number":1},{"id":"d7c783f4bfebd2055d7420d9e76ce5fb911441e3","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-09-23 17:38:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a66de2e99a60516e27e5e93ef9663f078984a245","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 01:33:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d5554ac5affec84c5c948bc838356dfa540de6ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 03:02:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8f46372596b40e18420c3a826043067 : SUCCESS in 4m 15s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0f2ad27b182449c3a620bf62b48ce8d8 : SUCCESS in 4m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/315aa07d138c4618883b29f9f56b8eaa : SUCCESS in 3m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5fcedd5560f74c059ffac14a7072e048 : FAILURE in 14m 49s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/63c6ef6086794046baf7159737441396 : FAILURE in 16m 58s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02f674cbaf6a40bea6f88c22b6d810d5 : FAILURE in 14m 24s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e4f271fa69d5419e872064ead8b638bd : FAILURE in 13m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3d945af9e3948d4ad3b9393d375f0b4 : FAILURE in 13m 29s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7be118401dfb49e6a6856540d3c1b9d3 : FAILURE in 7m 37s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e5f704dc5ac3462c97ea3137f8a47e7f : SUCCESS in 1h 26m 42s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ab1510bb9bf4e3ab9db7ac5b4510e4c : FAILURE in 24m 42s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ba479d4034744286ab3194a4e6387154 : SUCCESS in 27m 01s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46482aa193704fd597ad1fec29ce10ae : FAILURE in 24m 27s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b84762d0fc7747d7a8cf0b35e5b431e7 : FAILURE in 24m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907b11032d234a119d5365a8420bdc6c : SUCCESS in 4m 39s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/de323a46ba3d4e21bec5c41d87fa133a : FAILURE in 36m 34s","accounts_in_message":[],"_revision_number":2},{"id":"12d26579ac84ab5789fcaae4d986e16122ec8a21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-01 07:05:25.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"513a9b4e25bdb0d8f630c6c9033ab167dccf968b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 08:31:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29e13093741741d49a0604e3a3e830ea : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34ba1e2e44c844ffb76ffd905753b701 : SUCCESS in 5m 13s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2913311d98f4801bf0abb51529e1122 : SUCCESS in 5m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/6a71e14b37714ff3b3b4aea0e44bcacc : SUCCESS in 17m 53s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/e968d561a8d74e6d8266fbb4fa62ed34 : SUCCESS in 17m 56s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f3cd01fedb144c488871951d1a79b81 : SUCCESS in 18m 58s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0c2217ed5164b7a8802981d36208db5 : SUCCESS in 15m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d96f5ee3d57e43ab8fdda0cce7014a6b : SUCCESS in 19m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5a61996fcd4492d90f1110d91611140 : FAILURE in 11m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/7125040a1d5041ab96c5b303c485243a : SUCCESS in 1h 23m 33s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9d02446744a4b33827b37029743e3ba : FAILURE in 27m 51s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3c5674d3779543fd927f15cc4bacbde3 : SUCCESS in 35m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b41b842de4a645aa85687c984d3e8962 : FAILURE in 30m 12s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/918535b7b6c647aaa28c4d11d8a99742 : FAILURE in 27m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/160614ceafe64a3ebf1138463b79dd9c : SUCCESS in 5m 04s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/393451c436494ed799f778e207db4632 : FAILURE in 43m 49s","accounts_in_message":[],"_revision_number":3},{"id":"47cd485cd449cb43f7ff52aadccaa7ae890101bb","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-10-04 08:19:05.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"872379a18c82a0236c23e779266ae33158f164e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 09:50:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/878814c46c634fefa74ed1f0c09c4fc5 : SUCCESS in 4m 58s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cb00f99d65d44619f0e3bebe71eee19 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/612acd003655439992e7cbbfdb8b0dee : SUCCESS in 4m 48s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/3cf71667d2294716bc29406638fca38b : SUCCESS in 15m 11s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f9b533a31fd94790b3cdd2618d4f4c67 : SUCCESS in 19m 09s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d265974b69a404da592ec55c6cab257 : SUCCESS in 14m 18s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54701668181140de8c8bbf86c7641321 : SUCCESS in 14m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75133304565e46a78ef551c4b48f3f9e : SUCCESS in 12m 35s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/27fdb3e0b3674008b5906cd9989ea25d : FAILURE in 7m 53s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0d5ee7f113848649c5f43910484b2aa : SUCCESS in 1h 26m 02s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d13c36913db14107b403bdaed9666f2a : FAILURE in 24m 20s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/12b64ba1c56e4eedafa8dae645546787 : SUCCESS in 31m 27s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5af1509a087945d2bbbb5f6dd0049715 : FAILURE in 22m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4440dacc03774805ad30add1c76c20a3 : FAILURE in 23m 21s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/161b4d99a4c046269da72864b339ba39 : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/03cd605e326b440ab192eb5d461d0757 : SUCCESS in 1h 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"368aafc8cffe2b07d8dcb21bb6fc4da836dbcf44","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-10-04 13:43:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0f28d4526f240ede8b95a85d4434319c9116066a","author":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"date":"2021-11-04 11:53:39.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fb6a652a99157b638baa907435187fa68bb78a65","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-02-12 16:53:34.000000000","message":"Abandoned\n\nDone by https://review.opendev.org/c/openstack/puppet-swift/+/822269","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"76355f8a20df64b865380cb0e46abe77f7c3ecc6","revisions":{"1c575ad12e732d8b5fb758e4ab41dbbbb0c34ba4":{"kind":"REWORK","_number":1,"created":"2021-09-22 13:27:46.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/810433/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/33/810433/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/1"}}},"commit":{"parents":[{"commit":"ffbef544c89bc4733a4d5198af0bb65ccf6bd308","subject":"Prepare Xena M3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ffbef544c89bc4733a4d5198af0bb65ccf6bd308"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-22 13:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-22 13:26:32.000000000","tz":540},"subject":"Add support for [filter:slo] allow_async_delete","message":"Add support for [filter:slo] allow_async_delete\n\n... which was implemented during the Wallaby cycle[1].\n\n[1] e78377624ab7d34d2fedc06e907dcb9899d89b4a\n\nChange-Id: I75a75459b10d0dbf13f72694bc4a70eb4a958e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c575ad12e732d8b5fb758e4ab41dbbbb0c34ba4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c575ad12e732d8b5fb758e4ab41dbbbb0c34ba4"}]},"branch":"refs/heads/master"},"a6a362da3db0648bbd5d054e97a60489f4bd7264":{"kind":"REWORK","_number":2,"created":"2021-10-01 01:33:32.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/810433/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/33/810433/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/2"}}},"commit":{"parents":[{"commit":"9d03150ff3f167973477a826b451845e093b524c","subject":"Add information technology in the setup.cfg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/9d03150ff3f167973477a826b451845e093b524c"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-22 13:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-01 01:33:24.000000000","tz":540},"subject":"Add support for [filter:slo] allow_async_delete","message":"Add support for [filter:slo] allow_async_delete\n\n... which was implemented during the Wallaby cycle[1].\n\n[1] e78377624ab7d34d2fedc06e907dcb9899d89b4a\n\nChange-Id: I75a75459b10d0dbf13f72694bc4a70eb4a958e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a6a362da3db0648bbd5d054e97a60489f4bd7264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a6a362da3db0648bbd5d054e97a60489f4bd7264"}]},"branch":"refs/heads/master"},"76355f8a20df64b865380cb0e46abe77f7c3ecc6":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-01 07:05:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/33/810433/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/33/810433/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/33/810433/3"}}},"commit":{"parents":[{"commit":"5f57e5521a04ebd08fa68d6d7671528a535c14bb","subject":"Fix unit tests incompatible with puppetlab-stdlib 8.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5f57e5521a04ebd08fa68d6d7671528a535c14bb"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-09-22 13:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-10-01 07:05:25.000000000","tz":0},"subject":"Add support for [filter:slo] allow_async_delete","message":"Add support for [filter:slo] allow_async_delete\n\n... which was implemented during the Wallaby cycle[1].\n\n[1] e78377624ab7d34d2fedc06e907dcb9899d89b4a\n\nChange-Id: I75a75459b10d0dbf13f72694bc4a70eb4a958e49\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76355f8a20df64b865380cb0e46abe77f7c3ecc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76355f8a20df64b865380cb0e46abe77f7c3ecc6"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
