)]}'
{"id":"openstack%2Fpuppet-swift~816238","triplet_id":"openstack%2Fpuppet-swift~stable%2Fwallaby~I987a37cb0606b6c217f906fbae17a765e40f4580","project":"openstack/puppet-swift","branch":"stable/wallaby","topic":"bug/1949334-stable/wallaby","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-02 21:58:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987a37cb0606b6c217f906fbae17a765e40f4580","subject":"Fix missing argument of builder_file_path","status":"MERGED","created":"2021-11-02 00:44:08.000000000","updated":"2021-11-02 21:58:22.000000000","submitted":"2021-11-02 21:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816238-bug/1949334-stable/wallaby","cherry_pick_of_change":816237,"cherry_pick_of_patch_set":1,"meta_rev_id":"dea37bd42c9a97af6904dc3dbb4569d0f8999b5c","_number":816238,"virtual_id_number":816238,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 21:58:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-02 15:45:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-02 18:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-02 18:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 02:14:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:45:07.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-02 18:28:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"b327dd2fbda84d29aaf5933a91d68d82edffeb2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:08.000000000","message":"Patch Set 1: Cherry Picked from branch stable/xena.","accounts_in_message":[],"_revision_number":1},{"id":"e7de07dc6d7529ebb3c3b5a23c1e533e13d7276f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:16.000000000","message":"Topic bug/1949334-stable/xena-stable/wallaby removed","accounts_in_message":[],"_revision_number":1},{"id":"7d28482a73e565e899042765411a599de409e2d8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:21.000000000","message":"Topic set to bug/1949334-stable/wallaby","accounts_in_message":[],"_revision_number":1},{"id":"fd1e1a72b6f2902d40cfb2aebc007dc907ca458c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 02:14:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4388735c660433a95658fbd67bb496b : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f33926007b34011bfd7394d968b31a4 : SUCCESS in 3m 59s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a2801804e024ab2a2b0aaef649e5aad : SUCCESS in 14m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e67b1fed9f3a4ccb9904155ac52e0dd6 : SUCCESS in 12m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0cab4f367e7434bbf645f32842b8d8c : SUCCESS in 11m 13s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f938a5e27074a0f9933d8fde9e42e15 : FAILURE in 1h 04m 52s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/be9c12b33a854dd5945f942b5722f44c : SUCCESS in 1h 27m 53s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/9244b8f5803247f59a6994ca510c298f : SUCCESS in 28m 27s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe32d5ee7bb846c2a18d5b331f6e6497 : FAILURE in 24m 04s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/65541b80118343788892edc04df08d78 : SUCCESS in 1h 14m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6f753a683f0f37f9a14085cabdf789317237eade","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 15:45:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"096a53695f40d60a834f55f4b78b5a52635c0304","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:28:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6ab0a7f9b71c27e6b50eb3ad8214cef153a59e33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:35:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e8e85a712634c2ffe90b8d57db6cf1c10aaa8c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 20:14:38.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"77c802d3f682c2fb05dfdbd9371dcb759372df5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 21:58:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e59a5ca3d64d409e965f5f3ffb4749f8 : SUCCESS in 4m 38s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34cab81cfd094d90ae51d3792db69c3d : SUCCESS in 4m 44s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/402954bc385043a1ae2b7fd60516f7f0 : SUCCESS in 16m 21s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/921041a58e42410b8fdc1129451f9244 : SUCCESS in 13m 32s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/015909b64a8344fe87848aed2f953dc0 : SUCCESS in 1h 38m 25s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/55a711431af14078a059fe3bb68d28ce : SUCCESS in 33m 58s\n- puppet-openstack-litmus-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f1725e0e4014be1b6200f90de9cf396 : FAILURE in 31m 42s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/7ae3d187b99d4f6aad1027490e194812 : SUCCESS in 1h 20m 32s","accounts_in_message":[],"_revision_number":1},{"id":"dea37bd42c9a97af6904dc3dbb4569d0f8999b5c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 21:58:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8dc75e18323d8d1bf6e527fae3c5d056719c508","revisions":{"a8dc75e18323d8d1bf6e527fae3c5d056719c508":{"kind":"REWORK","_number":1,"created":"2021-11-02 00:44:08.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/38/816238/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/38/816238/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/816238/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/816238/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/38/816238/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/38/816238/1"}}},"commit":{"parents":[{"commit":"236a48239140a028ed12fcdb641734f960d3f5b2","subject":"Add system/project reader roles parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/236a48239140a028ed12fcdb641734f960d3f5b2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-01 07:43:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 00:44:08.000000000","tz":0},"subject":"Fix missing argument of builder_file_path","message":"Fix missing argument of builder_file_path\n\nThe change 63688a14e57282ae97072b406c81d0da2f61965c introduced support\nof storage policies and since the change was merged builder_file_path\ntakes one argument which is policy_index .\n\nThis change fixes the function call without argument to avoid failure\ncaused by missing argument.\n\nCloses-Bug: #1949334\nChange-Id: I987a37cb0606b6c217f906fbae17a765e40f4580\n(cherry picked from commit 350778570c04db749a61106f173419b25fd7494f)\n(cherry picked from commit e7d560738d5e02d32e63a0411211ddbb8afdcd9c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8dc75e18323d8d1bf6e527fae3c5d056719c508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a8dc75e18323d8d1bf6e527fae3c5d056719c508"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
