)]}'
{"id":"openstack%2Fpuppet-swift~816239","triplet_id":"openstack%2Fpuppet-swift~stable%2Fvictoria~I987a37cb0606b6c217f906fbae17a765e40f4580","project":"openstack/puppet-swift","branch":"stable/victoria","topic":"bug/1949334-stable/victoria","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-02 23:22:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987a37cb0606b6c217f906fbae17a765e40f4580","subject":"Fix missing argument of builder_file_path","status":"MERGED","created":"2021-11-02 00:44:41.000000000","updated":"2021-11-02 23:22:24.000000000","submitted":"2021-11-02 23:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816239-bug/1949334-stable/victoria","cherry_pick_of_change":816238,"cherry_pick_of_patch_set":1,"meta_rev_id":"6304ad5bf8039737b1cd5eecdc654e765041fb64","_number":816239,"virtual_id_number":816239,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 23:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-02 15:44:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-02 18:28:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-02 18:28:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 03:15:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:44:49.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-02 18:28:29.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"6c19c7efcde6c76c5b944d167af077cb0bf9f322","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:41.000000000","message":"Patch Set 1: Cherry Picked from branch stable/wallaby.","accounts_in_message":[],"_revision_number":1},{"id":"aa3b8b25dca5a54df15f59a507e6c389f58a908e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:47.000000000","message":"Topic bug/1949334-stable/wallaby-stable/victoria removed","accounts_in_message":[],"_revision_number":1},{"id":"c0e4a824ce035d72b6e5b259c8d1965b0b8c437b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:44:55.000000000","message":"Topic set to bug/1949334-stable/victoria","accounts_in_message":[],"_revision_number":1},{"id":"33c23d179ed7b4c80d9f5e4eef0062db090791e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 03:15:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4128e808cfaf415c8c9b5a97fade7483 : SUCCESS in 4m 50s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ad749360fd540bfa14a85f0331aa342 : SUCCESS in 8m 12s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/71c32cbb14d84d4ba387a933b00b2ff8 : SUCCESS in 14m 11s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b716b3ac0f7941a4aa68509eb69e8e82 : SUCCESS in 11m 35s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ada99eed070146f69d342a1c76337a1d : SUCCESS in 12m 44s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07b3adcd12f04fa9a86af93abf03da66 : SUCCESS in 41m 01s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0895053772fa4aa4968089974e718291 : SUCCESS in 1h 25m 02s\n- puppet-openstack-integration-6-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/465767a08af84ca4bfbe9d7d6ba3a5be : SUCCESS in 1h 50m 49s (non-voting)\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/53bb577df1b54321be322a4eb9b6c906 : SUCCESS in 26m 19s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/83e6bc3018584100bb4c0dc4d4259c88 : SUCCESS in 28m 41s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76134a90e7594ec98144c3758fdc5310 : SUCCESS in 35m 42s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2cb35f4d425841f68b733094191480f0 : SUCCESS in 2h 28m 18s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/b0ca2250ff104acbaf406c07e7894f9e : SUCCESS in 1h 12m 55s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/94dac74172d840c88e90c86f9a9136ce : SUCCESS in 1h 12m 44s","accounts_in_message":[],"_revision_number":1},{"id":"01f8d90f44eb1c449ffa0c8fe85d230b05d5f0db","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 15:44:49.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8e937090317567d33cabd91b5cc976f1457afd1d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:28:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0f93efcc1b079592765b09494f37b7d44215c64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:35:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7725d05baf7ed5b60d37f7492abfad70ce1587ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 20:14:49.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d17c463189ecdfa2a4075008065887cfda9c6502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:22:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6a39305a6fa0454586f44576986069aa : SUCCESS in 4m 52s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f6655cd0457d43fcad9785d6a52827ef : SUCCESS in 4m 23s\n- puppet-openstack-unit-6.14-centos-8-stream https://zuul.opendev.org/t/openstack/build/d99e9fccdbdb4f139983574ae6bfa944 : SUCCESS in 13m 40s\n- puppet-openstack-unit-6.14-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc277563d72b480e80f3596044b0c817 : SUCCESS in 13m 13s\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/76ae007614934dc39944493c24fbfb98 : SUCCESS in 1h 18m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/e6da34d3ad594d1abc0fd17bcd93f16d : SUCCESS in 1h 20m 46s\n- puppet-openstack-litmus-centos-8 https://zuul.opendev.org/t/openstack/build/a291b5849bf046a091b88ba5849ced9e : SUCCESS in 35m 18s\n- puppet-openstack-litmus-centos-8-stream https://zuul.opendev.org/t/openstack/build/67b008fc8f9444d183e3f622110e5186 : SUCCESS in 28m 43s (non-voting)\n- puppet-openstack-litmus-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/11df503820094523853a170c7ce84b8f : SUCCESS in 22m 45s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/755a760b47dd4a7abf1fe39ec10b3916 : SUCCESS in 2h 55m 41s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/db56ebd2796340308946189eaaafd97a : SUCCESS in 1h 22m 36s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8c936dd893e1415990a929625ffcd928 : SUCCESS in 1h 28m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6304ad5bf8039737b1cd5eecdc654e765041fb64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"131ac86b488021c27d597f4a1e8682684dfacb2f","revisions":{"131ac86b488021c27d597f4a1e8682684dfacb2f":{"kind":"REWORK","_number":1,"created":"2021-11-02 00:44:41.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/39/816239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/39/816239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/39/816239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/39/816239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/39/816239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/39/816239/1"}}},"commit":{"parents":[{"commit":"f74af484627e55ba10a333af27b296c045cbfa5e","subject":"Victoria and before: Build containers only once in TripleO jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f74af484627e55ba10a333af27b296c045cbfa5e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-01 07:43:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 00:44:41.000000000","tz":0},"subject":"Fix missing argument of builder_file_path","message":"Fix missing argument of builder_file_path\n\nThe change 63688a14e57282ae97072b406c81d0da2f61965c introduced support\nof storage policies and since the change was merged builder_file_path\ntakes one argument which is policy_index .\n\nThis change fixes the function call without argument to avoid failure\ncaused by missing argument.\n\nCloses-Bug: #1949334\nChange-Id: I987a37cb0606b6c217f906fbae17a765e40f4580\n(cherry picked from commit 350778570c04db749a61106f173419b25fd7494f)\n(cherry picked from commit e7d560738d5e02d32e63a0411211ddbb8afdcd9c)\n(cherry picked from commit a8dc75e18323d8d1bf6e527fae3c5d056719c508)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/131ac86b488021c27d597f4a1e8682684dfacb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/131ac86b488021c27d597f4a1e8682684dfacb2f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
