)]}'
{"id":"openstack%2Fpuppet-swift~816240","triplet_id":"openstack%2Fpuppet-swift~stable%2Fussuri~I987a37cb0606b6c217f906fbae17a765e40f4580","project":"openstack/puppet-swift","branch":"stable/ussuri","topic":"bug/1949334-stable/ussuri","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-02 23:26:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987a37cb0606b6c217f906fbae17a765e40f4580","subject":"Fix missing argument of builder_file_path","status":"MERGED","created":"2021-11-02 00:45:15.000000000","updated":"2021-11-02 23:26:53.000000000","submitted":"2021-11-02 23:26:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816240-bug/1949334-stable/ussuri","cherry_pick_of_change":816239,"cherry_pick_of_patch_set":1,"meta_rev_id":"1a8a78f5577f80b402b8156916dafbcaefb9a35e","_number":816240,"virtual_id_number":816240,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 23:26:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-02 15:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-02 18:28:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-02 18:28:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 03:16:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:44:45.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-02 18:28:30.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"f571b76aad8ff70f215a483b9a98991397774af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:15.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"269ee6058be4194fdcbb066d3cf994d05984e169","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:21.000000000","message":"Topic bug/1949334-stable/victoria-stable/ussuri removed","accounts_in_message":[],"_revision_number":1},{"id":"98e25df1df531ecded9f6ce074e289b0b6dd2ece","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:25.000000000","message":"Topic set to bug/1949334-stable/ussuri","accounts_in_message":[],"_revision_number":1},{"id":"a97ec08bdd6e7948d86a567f81817af823068857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 03:16:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b41d81eb916b4c9a8926df709561968f : SUCCESS in 4m 22s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/397b220731234455bd541653af1a3e85 : SUCCESS in 6m 07s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d77c406c686148e49c867d0e2d9d8b31 : SUCCESS in 3m 30s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/bcc4c65adced45f1b9c8ebee8002254d : SUCCESS in 11m 15s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/4e0710dc0e014fb99728fae4b9909575 : SUCCESS in 15m 42s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c79e8659e8134add8ae158848965a674 : SUCCESS in 9m 27s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/05d0047a6ad54d2fa6289574924e732a : SUCCESS in 12m 18s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e54697dbd199447e9e3a30557ceb3114 : SUCCESS in 40m 23s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/7bdf452f33b047dd8d68abca78de57fe : SUCCESS in 1h 01m 59s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e3427bc7c50343c18d64ff98b4c1c177 : SUCCESS in 58m 33s (non-voting)\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/9c0ac4dce1ef4328a8d3efa1ef6f9564 : SUCCESS in 38m 58s (non-voting)\n- puppet-openstack-beaker-centos-8-stream https://zuul.opendev.org/t/openstack/build/c8f9933f8c924a8eaecfc8027785c06c : SUCCESS in 27m 38s (non-voting)\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c051c64e679a41eb84d7bc78e2715cc8 : FAILURE in 6m 37s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/d274f83a648245929039dabb08826289 : SUCCESS in 2h 28m 55s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/3db6a259a87e4a1a8db8752122246847 : SUCCESS in 1h 15m 25s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/79c070e003064dd1be6d7c87e99a32aa : SUCCESS in 1h 11m 43s","accounts_in_message":[],"_revision_number":1},{"id":"31fe19ce7c321c76394e73cc5e6cccb41ab8e75c","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 15:44:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b289e03e7f204279908cb2a59aa59b8037e8c0d","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:28:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5df3f8b3072cec0db085a78f57a58db39bd7382b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:34:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b57422ec6e172fed0257ab53af975c12a78f2dd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 20:15:33.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ac9735c65fd0e1b6b6d59e671952944b2dd0b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:26:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/86eccb830abb49d689f6d62c7613d19e : SUCCESS in 4m 39s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ca26e74b4d9f4d0a84105d6c110f51ef : SUCCESS in 4m 14s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/450bdafda337463a90b88b3050cb2d92 : SUCCESS in 4m 43s\n- puppet-openstack-unit-5.5-centos-8-stream https://zuul.opendev.org/t/openstack/build/e02edddfa3b24bce91359ce3c579d0f3 : SUCCESS in 17m 44s\n- puppet-openstack-unit-6.11-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a0b728ed7e34e27b08af6477805338f : SUCCESS in 17m 34s\n- puppet-openstack-unit-6.11-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d3ec64f9985a412bb0fc54a7beddc766 : SUCCESS in 12m 41s\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/632bdd475d8847a09d67663565fe70b5 : SUCCESS in 51m 17s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/0fb24c4df42b46b7994280cbb752f950 : SUCCESS in 1h 06m 41s\n- puppet-openstack-beaker-centos-8 https://zuul.opendev.org/t/openstack/build/616a1a6291f04f66803afaee85a16ebc : SUCCESS in 34m 18s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/8374c7fbd5c946e5940df00511488152 : SUCCESS in 3h 00m 31s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/7275e7250a7f4e5f9b2f45383f40bb9d : SUCCESS in 1h 30m 29s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fd5e057d96354eabb92425730efed923 : SUCCESS in 1h 20m 59s","accounts_in_message":[],"_revision_number":1},{"id":"1a8a78f5577f80b402b8156916dafbcaefb9a35e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:26:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"edfaa3546a941467dcee9a0f268e1a90656b4526","revisions":{"edfaa3546a941467dcee9a0f268e1a90656b4526":{"kind":"REWORK","_number":1,"created":"2021-11-02 00:45:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/40/816240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/40/816240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/816240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/816240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/40/816240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/40/816240/1"}}},"commit":{"parents":[{"commit":"ddb37d9e41f850056830a0d4f562f61c00547eca","subject":"Victoria and before: Build containers only once in TripleO jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ddb37d9e41f850056830a0d4f562f61c00547eca"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-01 07:43:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 00:45:15.000000000","tz":0},"subject":"Fix missing argument of builder_file_path","message":"Fix missing argument of builder_file_path\n\nThe change 63688a14e57282ae97072b406c81d0da2f61965c introduced support\nof storage policies and since the change was merged builder_file_path\ntakes one argument which is policy_index .\n\nThis change fixes the function call without argument to avoid failure\ncaused by missing argument.\n\nCloses-Bug: #1949334\nChange-Id: I987a37cb0606b6c217f906fbae17a765e40f4580\n(cherry picked from commit 350778570c04db749a61106f173419b25fd7494f)\n(cherry picked from commit e7d560738d5e02d32e63a0411211ddbb8afdcd9c)\n(cherry picked from commit a8dc75e18323d8d1bf6e527fae3c5d056719c508)\n(cherry picked from commit 131ac86b488021c27d597f4a1e8682684dfacb2f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/edfaa3546a941467dcee9a0f268e1a90656b4526"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/edfaa3546a941467dcee9a0f268e1a90656b4526"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
