)]}'
{"id":"openstack%2Fpuppet-swift~816241","triplet_id":"openstack%2Fpuppet-swift~stable%2Ftrain~I987a37cb0606b6c217f906fbae17a765e40f4580","project":"openstack/puppet-swift","branch":"stable/train","topic":"bug/1949334-stable/train","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-11-02 23:26:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I987a37cb0606b6c217f906fbae17a765e40f4580","subject":"Fix missing argument of builder_file_path","status":"MERGED","created":"2021-11-02 00:45:43.000000000","updated":"2021-11-02 23:26:54.000000000","submitted":"2021-11-02 23:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816241-bug/1949334-stable/train","cherry_pick_of_change":816240,"cherry_pick_of_patch_set":1,"meta_rev_id":"7b068c319cdb637cd4571ce1e31122a8c7973fff","_number":816241,"virtual_id_number":816241,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-02 23:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":2,"date":"2021-11-02 15:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":2,"date":"2021-11-02 18:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":1,"date":"2021-11-02 18:28:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 03:00:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-02 15:44:53.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"},{"updated":"2021-11-02 18:28:31.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"95987aefa31a19ca8ac3f0ea0e8d6092f22a901d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.","accounts_in_message":[],"_revision_number":1},{"id":"611ed495f31b50cf4234108b4e118c8a0bc575a7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:50.000000000","message":"Topic bug/1949334-stable/ussuri-stable/train removed","accounts_in_message":[],"_revision_number":1},{"id":"c928c1f6cbeaea42d6be4471afa18723b610a923","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-02 00:45:55.000000000","message":"Topic set to bug/1949334-stable/train","accounts_in_message":[],"_revision_number":1},{"id":"9053c079d040ecce5f63b225adc6c1f1fe748316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 03:00:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e6c6a8d419ac459caf426a908c7efb47 : SUCCESS in 4m 53s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/90ea7da30b1a4b13abb5ca8e866fce84 : SUCCESS in 4m 31s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbc0204f6f25434984dad8a5b2455ae1 : SUCCESS in 6m 01s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/7c885d766d8246cf82a8403134eb4366 : SUCCESS in 17m 00s\n- puppet-openstack-unit-5.5-centos-8 https://zuul.opendev.org/t/openstack/build/58c3077d12154ba0abfc386f672c6bcf : SUCCESS in 11m 49s\n- puppet-openstack-unit-6.4-centos-8 https://zuul.opendev.org/t/openstack/build/fb12057d6146455587687f36ae74a52a : SUCCESS in 13m 16s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/566a2f4f23c44cd1bb666a996db64852 : SUCCESS in 14m 50s\n- puppet-openstack-unit-latest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/35de4d2d1a314314a03eaf0eb4e721ea : SUCCESS in 11m 44s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/43844fbad5884e609b7a41fa5589be29 : SUCCESS in 41m 55s (non-voting)\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/cd8872c085844ba99a43d95d4717d974 : SUCCESS in 1h 04m 28s\n- puppet-openstack-integration-5-scenario002-tempest-centos-8 https://zuul.opendev.org/t/openstack/build/4a2dab638ce74324a516967017315d01 : SUCCESS in 45m 20s (non-voting)\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/a495768cc5104d7895f6759f4ce955b6 : SUCCESS in 28m 59s\n- puppet-openstack-beaker-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/67545d632c50491b9a13ecdd5e119040 : FAILURE in 8m 12s (non-voting)\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/2ec0f75644d64b91b24b5138441fcf24 : SUCCESS in 2h 04m 15s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/59f5572e3071474ca82e00216094a7b0 : SUCCESS in 1h 15m 45s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/78d7cd05415349c88fe39a224157b813 : SUCCESS in 2h 13m 24s","accounts_in_message":[],"_revision_number":1},{"id":"10e11c0ccd9982a164cca2d3c4d7035b92f938a6","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-11-02 15:44:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"594182370bfbea46230e1c21c825c6a97f50d161","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-11-02 18:28:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"476a90d440e1449dfe9278b4dd38e4c7207db0a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 18:35:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbae8232cf4c59858647b6a206e2d5fb08e56a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 20:16:18.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fee9016bfa31195cfa4bd621b85ae68c8abba045","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:26:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6ecda2c68441454db50df0079b41ee4a : SUCCESS in 7m 01s\n- puppet-openstack-syntax-5-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/631c558613224436ba5b8ac3a34b0839 : SUCCESS in 5m 30s\n- puppet-openstack-syntax-6-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/04e00e8a9a744c5bba02cd513cfbc448 : SUCCESS in 5m 46s\n- puppet-openstack-unit-5.5-centos-7 https://zuul.opendev.org/t/openstack/build/25ee556a1b4d4912a57e587f3bb71a0f : SUCCESS in 24m 32s\n- puppet-openstack-unit-6.4-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c39bdb8cf51b4f2a8f331ffa07a62575 : SUCCESS in 13m 15s\n- puppet-openstack-integration-5-scenario002-tempest-centos-7 https://zuul.opendev.org/t/openstack/build/1bacc283a23b4e59945463549603719c : SUCCESS in 1h 19m 35s\n- puppet-openstack-beaker-centos-7 https://zuul.opendev.org/t/openstack/build/58de33ff99984328992c51929f43d745 : SUCCESS in 28m 16s\n- tripleo-puppet-ci-centos-8-content-provider https://zuul.opendev.org/t/openstack/build/fed47956169a41af9ec0787edbae56ec : SUCCESS in 2h 55m 57s\n- tripleo-puppet-ci-centos-8-undercloud-containers https://zuul.opendev.org/t/openstack/build/5eefb2090f1a4650a8dbcc5448b95d0d : SUCCESS in 1h 54m 41s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ddb02375b8304115bb122f2955dec0d0 : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"7b068c319cdb637cd4571ce1e31122a8c7973fff","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 23:26:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1c1cb38c701bcdec161ebbe66d5913f99359598b","revisions":{"1c1cb38c701bcdec161ebbe66d5913f99359598b":{"kind":"REWORK","_number":1,"created":"2021-11-02 00:45:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/41/816241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/41/816241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/41/816241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/41/816241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/41/816241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/41/816241/1"}}},"commit":{"parents":[{"commit":"44f6d8b1a5cf9d6e63f864e9b82ba1046d7f7453","subject":"[Train only] Run TripleO jobs on CentOS8 instead of CentOS7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/44f6d8b1a5cf9d6e63f864e9b82ba1046d7f7453"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-01 07:43:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-02 00:45:43.000000000","tz":0},"subject":"Fix missing argument of builder_file_path","message":"Fix missing argument of builder_file_path\n\nThe change 63688a14e57282ae97072b406c81d0da2f61965c introduced support\nof storage policies and since the change was merged builder_file_path\ntakes one argument which is policy_index .\n\nThis change fixes the function call without argument to avoid failure\ncaused by missing argument.\n\nCloses-Bug: #1949334\nChange-Id: I987a37cb0606b6c217f906fbae17a765e40f4580\n(cherry picked from commit 350778570c04db749a61106f173419b25fd7494f)\n(cherry picked from commit e7d560738d5e02d32e63a0411211ddbb8afdcd9c)\n(cherry picked from commit a8dc75e18323d8d1bf6e527fae3c5d056719c508)\n(cherry picked from commit 131ac86b488021c27d597f4a1e8682684dfacb2f)\n(cherry picked from commit edfaa3546a941467dcee9a0f268e1a90656b4526)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c1cb38c701bcdec161ebbe66d5913f99359598b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1c1cb38c701bcdec161ebbe66d5913f99359598b"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
