)]}'
{"id":"openstack%2Fpuppet-swift~821646","triplet_id":"openstack%2Fpuppet-swift~master~I752dc30755febcfa94cb266c2faf3b7aedc35ca9","project":"openstack/puppet-swift","branch":"master","topic":"upstream","attention_set":{},"removed_from_attention_set":{"26721":{"account":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"last_update":"2021-12-20 17:18:19.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-20 07:19:10.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"I752dc30755febcfa94cb266c2faf3b7aedc35ca9","subject":"Add rsync parameters for object server","status":"MERGED","created":"2021-12-14 04:38:40.000000000","updated":"2021-12-20 17:20:36.000000000","submitted":"2021-12-20 17:18:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"821646-upstream","meta_rev_id":"b3c7b4278bc13d6e92a0228ad8cca2d2c3bebafd","_number":821646,"virtual_id_number":821646,"owner":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 17:18:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":2,"date":"2021-12-20 07:19:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2021-12-20 15:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2021-12-20 15:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-14 05:33:08.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2021-12-14 05:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:31:44.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"}],"messages":[{"id":"8c64d4f292541fb444c027e9ef92f4c8d425a08f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-14 04:38:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f99cb1b5145ec099baf0b2a995aa504e583d24a8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-14 05:33:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f7436ce59b1561f1fa8214a9248886b4b1512914","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-14 05:54:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec7a2569f3a0458aa309a58183abb120 : FAILURE in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/72cbb705bb114b72965fa8e147cc22ed : SUCCESS in 3m 29s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9dd5bc3ba8bf4f1992c2b0477e3adfc9 : SUCCESS in 3m 41s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/db3f1e51b94a4e079b1edea59c8c28a0 : SUCCESS in 15m 23s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/15cfeeabc640489b8646c2e293a99f7c : SUCCESS in 14m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d4d51db4a2ef4e659ba825f7dea5113d : SUCCESS in 11m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7181fa0e6ffc490981746b2aa232bca3 : SUCCESS in 14m 42s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/88b607e1d0984cf19e7eef8d0a709387 : SUCCESS in 11m 22s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c1ded5a0a184e33a7d423fddd445ed6 : FAILURE in 35m 58s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/17e37e90521f4849bc66a41973d6deea : FAILURE in 59m 25s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/93574bfcffea4308812241d029da88a2 : FAILURE in 38m 17s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23222a03ada64b28bc20c9ad946d570d : FAILURE in 25m 07s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/03756d64acbe44a185099f9b6113abca : SUCCESS in 26m 45s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c61f9b07e7de4f80b1c2a7e6e41ec0b6 : FAILURE in 25m 33s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35fb9fc337d540929d46946a32e20854 : FAILURE in 24m 45s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/18e162ec46174a5eb21f129e814e8969 : FAILURE in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"930871179694a5e8c55c058b68b986b1d9342e6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-15 00:06:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6dddfdea7354331f759deeb553bcda627a4838b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 01:32:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b9b44eed9db453295afad379e465ce8 : SUCCESS in 3m 24s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9cb395d08e6a451aa1d16f44c2b788aa : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/124b8e7485374f5db0de09fbd5c15b46 : SUCCESS in 3m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a7200743d71b4f699d340b77bd666dd0 : FAILURE in 16m 39s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/93e8eb7a4a6447efb8ce28e8fc76b087 : FAILURE in 12m 44s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7dfefec7fd11404e84fb8b65d1ad691d : FAILURE in 12m 10s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0996ee5f9c994858b80d689e93540c38 : FAILURE in 12m 38s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c86cfe52d8074dd1b22810ab03101f56 : FAILURE in 10m 20s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96f05f469cfc44cab7b12b94da49bce3 : FAILURE in 37m 10s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/990ee17f4c784e16aeb94beb132ac981 : FAILURE in 48m 06s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/104dae421ffc4dd6b9bd370ed1219429 : FAILURE in 45m 02s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b20f2055048a4b7da1939aebadb58fc9 : FAILURE in 22m 52s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/853efc3430df426cb70a472f8ec9e143 : SUCCESS in 33m 32s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/119964ba3a3c424bbc8e554e044dc3fc : FAILURE in 26m 39s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7e698c71a7a48079dadfe258e8a92d9 : FAILURE in 23m 28s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fff65f181a534ced90ed9cadf52bc603 : SUCCESS in 4m 11s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/548a523d86d944cdaf4811b427d60c67 : FAILURE in 1h 23m 38s","accounts_in_message":[],"_revision_number":2},{"id":"5f0e01e8ad98ff258111a8752643ce083f34dd78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-15 02:01:08.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9f3411f96356e1ae7fe4dfc97af726d85baa6ce0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-15 02:01:18.000000000","message":"Topic upstream removed","accounts_in_message":[],"_revision_number":3},{"id":"b5deec56f0c449b5abc1dfbaafa732dbc8b98514","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 03:38:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8bc285b241204ae88597673515efa3f4 : SUCCESS in 5m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a580c003aa54dc2a7ef857fbc26094e : SUCCESS in 5m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17298c316d0f46c6b6fe6a3833a153ca : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/ff4b287080824271aceb29e4cfda3c07 : SUCCESS in 21m 00s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/82c700dd12db48ceb08b601731d142cc : SUCCESS in 19m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02d2574b879c4d2ea8d1305439abd61d : SUCCESS in 19m 27s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c43d60a48be499080063780e6604d43 : SUCCESS in 17m 27s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01ab3e8fade9429eb1c4e4f5ac894925 : SUCCESS in 16m 17s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/606442e1ba3e4e59adfcbc923c729820 : FAILURE in 41m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/bbaa023556af40159db20c9b9c859be1 : FAILURE in 55m 50s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bea8817f2f241f4bc13f09a8fcd4e14 : FAILURE in 42m 03s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/924779d29978419a9528719fd09e738f : FAILURE in 29m 38s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d3724cb288434b6f9a10d362163912e4 : SUCCESS in 32m 37s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/75b9a1f370e944d99f202aaa57a1b9f0 : FAILURE in 21m 09s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97bbeff5ae61448292e5c2d001a5d69f : FAILURE in 25m 49s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e63a71793add49399e508d9cc2382bdb : SUCCESS in 5m 53s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f7c486d253c64e32973a4ac1dbb9252c : FAILURE in 1h 24m 15s","accounts_in_message":[],"_revision_number":3},{"id":"5bedab68e72ae180b559b6fd887e0555f3da1115","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-15 03:57:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3e5ad9efdaebfff265af2fa7175a7dd7aff83049","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-15 05:23:07.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb92a17d79e341e9bddcc3f249676977 : SUCCESS in 3m 39s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d41718b2259347b28fa074cd61bd5f68 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6216fa186e6b473385737b0fa47c37f5 : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c764bc06bfca4d09a85f520a68da5bbd : SUCCESS in 16m 59s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1c660a4459c74f8c87c334bed86d44fb : SUCCESS in 14m 35s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ebc91de99e84f1286aa32008c4521c4 : SUCCESS in 13m 35s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7bdd2306009b4a8a84f151000a914cf2 : SUCCESS in 13m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f17d4e10855b4ad1ad9aac31cfc35157 : SUCCESS in 12m 55s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/02975c10e22641ccbdb5afbe8ce78694 : FAILURE in 32m 07s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa19fc1f096640cfae6f6e4e4f42d3cd : FAILURE in 47m 08s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c62fcec9099c4ec396b941068e4ff6e2 : FAILURE in 39m 03s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/660d5a1750f44c2185c54fb64e80b960 : FAILURE in 22m 54s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f3a301dc488642ac968bc8f828558280 : SUCCESS in 27m 55s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d90c6aec63e4db48da33dd34a46062b : FAILURE in 25m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9466f7af06bc42f792e29c0fedb625d7 : FAILURE in 28m 07s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/864975dc27f04739b4a268cd7d5a5692 : SUCCESS in 3m 19s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/3386b5f9932f41ab989e9bfe38181e6e : FAILURE in 1h 20m 09s","accounts_in_message":[],"_revision_number":3},{"id":"efb0ea43d744ce4f291eaf5e2a4b81546a470077","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-15 23:15:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b52e67efa7b1f95cca4e338401b9d24cb06667b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 00:39:38.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/21b737c3ff29410fa7d38284f132022b : SUCCESS in 4m 05s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70185276172f4bdfa74e6fb21affb349 : SUCCESS in 4m 22s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/33684bee1f1742ea8ceedf31009ace8f : SUCCESS in 3m 54s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/bd07139242964811ad2000ddaa11a90b : SUCCESS in 17m 39s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dba5a30bf5514ad1a220cb5d3946a7d5 : SUCCESS in 17m 13s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aaf483926be54429820966801be4a42d : SUCCESS in 15m 08s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96869288c0154969a4e503f48bd13877 : SUCCESS in 15m 53s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/605cd4ba454e4990957004cc58f25211 : SUCCESS in 14m 56s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6556ac506b114a7bb322be3e9548ba2c : FAILURE in 42m 59s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/73a9cfb315304f2d958d09443feb038c : FAILURE in 54m 12s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/4785329bbfb84c3195f07429913178ce : FAILURE in 41m 15s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/788d84ca8a35472286fc81461e87ac8f : FAILURE in 26m 08s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/76f8981d82b942e5be99c04d35d7eaf7 : SUCCESS in 32m 58s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f89683cb50dd46f2858fd6cbf9cc7a04 : FAILURE in 25m 25s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0fd24702ea142de98290fdec62c0bb1 : FAILURE in 27m 47s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c3e477ae32a42c3890a369cec5c80fc : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/bcaf9e63f3244a6aa766b024c9c6ef8c : FAILURE in 1h 19m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e82dd0a560f00105e28d83a0a5e2918293f9f9e6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-16 02:12:07.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86269336a53d88f6a4abdd17421b9e3fe9cd2a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 03:45:49.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b80a5c51f604aeea796b5eed1d62617 : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8ec31a061d24b718d190a7e077113b6 : SUCCESS in 4m 06s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1dbfe61b451f4fb4aabd7198c59f39e0 : SUCCESS in 4m 38s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aaf778d6c13643b1af747db6f0b700a5 : SUCCESS in 18m 08s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f2798b8246e7416ab805bfa7982c4e1f : SUCCESS in 17m 15s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53128b4d94604d09a96a5ee92d4b2c48 : SUCCESS in 16m 20s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/81310162504f4298a742a09df98b329f : SUCCESS in 15m 54s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/184972a122cc484bbd4b6fd81167e223 : SUCCESS in 14m 51s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09c1b8fbf2514ee297aa5d697f6438e3 : FAILURE in 39m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/0c6f5c9469f742eda9f44a1bda50ee68 : FAILURE in 47m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/26c11d36149e4633bc94174489d57344 : FAILURE in 41m 57s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78e44a1364904f6b82100b7250f4d6d4 : FAILURE in 23m 50s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cf93e34718c74d618ec4684ee5c9390b : SUCCESS in 28m 03s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8a33f01d8cf74c38816b9e64bf69dadf : FAILURE in 32m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f186a6b52d1d43328f958b9dfb3cd114 : FAILURE in 33m 11s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01d83d81111b446ca04bcd9cc5b99587 : SUCCESS in 5m 01s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ec54091020b54c5eaccefe5cdea321f8 : FAILURE in 1h 08m 48s","accounts_in_message":[],"_revision_number":3},{"id":"18a1fb36792ba06b2ef3dc84e1c1b85b90276ba1","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-16 06:02:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7df8d5ba48ce1af12bbeb218c678cf7e8a689fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-16 07:08:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f52773f04962497aa857c2d7a353c5da : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/189cdf53ddca48ddb4cd8e05c99d94d1 : SUCCESS in 3m 55s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f25550b639642728556e720538e10bf : SUCCESS in 3m 07s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/77a3a13df45e4b8aadfac3bf75c799c7 : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fda2462c9b14a5e916e8e8ab199c9b0 : SUCCESS in 12m 45s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/043072d297d24308bfa0c07546ecb304 : SUCCESS in 11m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab71458ff25a40bea4e5a4fd1e3495fe : SUCCESS in 13m 41s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bca7a26eb00c4526896d8b3a73662306 : SUCCESS in 14m 12s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c057408acee47c38253cce673b9f1cc : FAILURE in 37m 04s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/b8db6a7b9d1740c98c049c252d810cce : FAILURE in 39m 10s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0076bace4af46349242b3f556d629e0 : FAILURE in 31m 55s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c812693ceac4b96bd311c36c3e88e45 : FAILURE in 25m 50s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/52dc54511ff449faa3c2087b4819ef55 : SUCCESS in 59m 03s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5af758d09a7e4a71a55682084dda0ee4 : FAILURE in 21m 04s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ded9131b2c29448d86227b4d2102af5b : FAILURE in 23m 58s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbd804914ff54150ab7431f801ca0121 : SUCCESS in 3m 30s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/99540ec9e847416d95fd23a3850b011f : FAILURE in 1h 03m 30s","accounts_in_message":[],"_revision_number":3},{"id":"f78ea5d236011b3d49f3b85e8a3f32f3c3c7d783","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-17 04:20:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1fdf9a5ad64ab5faf574f4b0f696c8c16f7d81c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 05:21:33.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/253b341c5aa649e399a729e8fc86fb5d : SUCCESS in 4m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e053b9e07c9a482382573a3737033096 : SUCCESS in 4m 24s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ae3509f9d1541f2a5a4223298d0dc84 : SUCCESS in 3m 46s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/b9fc24385e08444aa1c3069ef0315fb2 : SUCCESS in 12m 36s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2f7e0af1c8de42d998de10e977b08198 : SUCCESS in 13m 11s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5380562983774131996d1b5c3993034b : SUCCESS in 12m 03s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d58b45b7cd364b9c80bb068849084f86 : SUCCESS in 12m 39s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a8be81f77a1c495bb2236c4cac53c060 : SUCCESS in 11m 50s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ad63b2c6a0e4950ae818b00f5fb5a6f : FAILURE in 30m 23s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2938b6c5d2b465b954b58ff89b03105 : FAILURE in 37m 52s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cb1ecc7d5d749ae82a44a4c1e8698f7 : FAILURE in 43m 51s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b00271250c284ae593373f70deedd7f9 : FAILURE in 20m 39s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/de6472ad14e647909005c692f47a545d : SUCCESS in 23m 54s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce0eaa1117e7427394bec05f6c6caa80 : FAILURE in 19m 10s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/511b195778814568b9a33b798753f2ff : FAILURE in 22m 37s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9310c4951958438e82a061f19093ce9f : SUCCESS in 3m 34s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/69d446b0afc64cce9e48e790cd4b69a7 : FAILURE in 57m 35s","accounts_in_message":[],"_revision_number":3},{"id":"34ef657d8938e0b07e525beab6371e5460df63b9","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-17 05:30:03.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3356db55f9145df490ec9262c49bafe7ca43d314","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-17 05:43:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6b20f21ecaeef08748a6459b62cdcb552d57f8d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-17 05:48:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"85eee04d09c4ef96a05d15581be522ef2c5682e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-17 07:25:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b2a88c94f3a4e03a236a03f33119837 : SUCCESS in 3m 04s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f460d7ee0114144a792d171b33dc590 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29c30017772941bcbe0d55af5f8f9436 : SUCCESS in 2m 47s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/74a6edb0e5b14b1fa629cbd756ee07c8 : SUCCESS in 16m 05s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a24e274166e645dcac98c2f4cf08d6ca : SUCCESS in 13m 26s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2454a2773ca34be79275985689ef77d3 : SUCCESS in 13m 50s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a03ccbff0bec48b48cab9ddae3619020 : SUCCESS in 15m 03s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/492e177f96784e3a8d764c5b0324a9c7 : SUCCESS in 13m 38s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d61c0f5d7ea45128218f57bf9c5981a : FAILURE in 36m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8b5f4381d45847739d847d35060ea524 : SUCCESS in 1h 31m 18s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3239940e83745f3a6f31b390e865bed : SUCCESS in 1h 08m 50s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c8acd746c0c4032911d6fd422aeb392 : FAILURE in 17m 45s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b561ddc7053e464b9ed7acd6b2c0bfd8 : SUCCESS in 30m 33s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/95b08902a1514982b85c53270ab2736a : FAILURE in 23m 00s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3981bf2eae5f4faea68b2c2c297fcd0b : FAILURE in 17m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/984dd96ef5d24243a91c33bceba0a03c : SUCCESS in 3m 25s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/d00d3a970da14968b65396f822a04c7e : FAILURE in 1h 15m 18s","accounts_in_message":[],"_revision_number":4},{"id":"fc47bd8bb1bfecc54d7e9e19ce3b0327a607eb58","author":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"date":"2021-12-19 22:29:01.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e05932ed5382fa0de1f96e8ac541c1a995d8521","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 23:43:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/53f10ff75cbf4c069060369e1255271c : SUCCESS in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91e56f551496423096805b8f3250cd8a : SUCCESS in 5m 09s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d8b18e336e04a97aa1257ba7b6fa4f1 : SUCCESS in 6m 42s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/eb56eea08c3142118d6b8d91d6b5b6c0 : SUCCESS in 12m 30s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1d98ca602b0a454489927e925c04bed2 : SUCCESS in 12m 19s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0781cd79199642bba7b2230063d7f4a1 : SUCCESS in 10m 30s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/92a11e7f2816483fb9caef540f5e695c : SUCCESS in 10m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7558e87562b49b0924bd12186e1791e : SUCCESS in 9m 33s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b147c1d10dc34d8db4a525f8c71e5f0f : FAILURE in 28m 01s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/31e61ea2674445fdaefd2c8c85a9e6ba : SUCCESS in 1h 02m 29s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/111fe3c3c3a947058bfa97c41a09bccd : SUCCESS in 48m 03s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f8b0c640a1d47ff8274b459eea5e1fd : FAILURE in 18m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dd862196c44d41dd91d2d634765d450f : SUCCESS in 22m 43s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9945ab170ef4f3090e960fe458f2cca : FAILURE in 23m 24s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3a2ea6bbcdf49acacbfbd271df5e8d0 : FAILURE in 26m 39s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9d171c16277499ba4a93b424160b1c3 : SUCCESS in 3m 18s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/e6f6fe344aa34732b8863b228cdd95fa : SUCCESS in 1h 10m 12s","accounts_in_message":[],"_revision_number":4},{"id":"8bbea98358afd760d2f67c3a2768cd5243a85a8d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 07:19:10.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"18e52a01bc08e3c9ae9e610d9f59822f9165738f","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:31:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"a843081693a48da1e56d23d47235cbc72b9a87bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 15:33:19.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"17934db371b97ec169fcc058c0fc9866466e35e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 17:18:18.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24f08be0fb7e4f708a56f8a3080ed5d9 : SUCCESS in 3m 18s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4cb04442e6c41eb928ff7e42af116e8 : SUCCESS in 3m 31s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69328e1d7b86446b8d0cad7f3c509e0b : SUCCESS in 3m 13s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/da0dda9543994b878fa1c7965f9aeee7 : SUCCESS in 14m 44s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/acd71b2c80df451db1cee438ab4ff56d : SUCCESS in 15m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8b0b1a11b430470caad8619eff3ea0ca : SUCCESS in 15m 02s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/611c1f37b1fa4fe38054fe970918171f : SUCCESS in 15m 37s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/acfbd702b9c9497ca47c5f09d2233c52 : SUCCESS in 1h 40m 58s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/a402c0c1110940e2848038a92a48b0de : SUCCESS in 31m 39s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f50fa393388e4870b503c53e65eb9abd : FAILURE in 25m 29s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f298ec25b89e4cd098101b1268719469 : SUCCESS in 3m 50s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/aae1b7f6b47b4a41941482dc8fe48c74 : SUCCESS in 1h 18m 22s","accounts_in_message":[],"_revision_number":4},{"id":"0971f15e805adb0dc789593b98d661c0d3057795","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 17:18:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"b3c7b4278bc13d6e92a0228ad8cca2d2c3bebafd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 17:20:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c0e629cc3f2438793c361fdfb775ddc : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"54c75b9f1f72f927d10edbaab7ddde4934b5e7e5","revisions":{"76bc6185ca61b7c1c12400a15cdaf70cd61ca58c":{"kind":"REWORK","_number":1,"created":"2021-12-14 04:38:40.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/46/821646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/821646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/1"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-26 12:07:30.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-12-14 04:20:49.000000000","tz":660},"subject":"Add rsync parameters for object server","message":"Add rsync parameters for object server\n\nChange-Id: I752dc30755febcfa94cb266c2faf3b7aedc35ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76bc6185ca61b7c1c12400a15cdaf70cd61ca58c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/76bc6185ca61b7c1c12400a15cdaf70cd61ca58c"}]},"branch":"refs/heads/master"},"44d4b6eaf515f3a15bee20c1eb02184d49b653ce":{"kind":"REWORK","_number":2,"created":"2021-12-15 00:06:52.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/46/821646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/821646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/2"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-26 12:07:30.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-12-15 00:06:32.000000000","tz":660},"subject":"Add rsync parameters for object server","message":"Add rsync parameters for object server\n\nChange-Id: I752dc30755febcfa94cb266c2faf3b7aedc35ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/44d4b6eaf515f3a15bee20c1eb02184d49b653ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/44d4b6eaf515f3a15bee20c1eb02184d49b653ce"}]},"branch":"refs/heads/master"},"3347aedd99fa482f5d12e5654536b539801fa0b8":{"kind":"REWORK","_number":3,"created":"2021-12-15 02:01:08.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/46/821646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/821646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/3"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-26 12:07:30.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-12-15 02:00:58.000000000","tz":660},"subject":"Add rsync parameters for object server","message":"Add rsync parameters for object server\n\nChange-Id: I752dc30755febcfa94cb266c2faf3b7aedc35ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3347aedd99fa482f5d12e5654536b539801fa0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/3347aedd99fa482f5d12e5654536b539801fa0b8"}]},"branch":"refs/heads/master"},"54c75b9f1f72f927d10edbaab7ddde4934b5e7e5":{"kind":"REWORK","_number":4,"created":"2021-12-17 05:48:52.000000000","uploader":{"_account_id":26721,"name":"Rocky","display_name":"Shi Yan","email":"yanshi.403@gmail.com","username":"Rocky"},"ref":"refs/changes/46/821646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/46/821646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/46/821646/4"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2020-05-26 12:07:30.000000000","tz":600},"committer":{"name":"Rocky","email":"shi.yan@unimelb.edu.au","date":"2021-12-17 05:48:32.000000000","tz":660},"subject":"Add rsync parameters for object server","message":"Add rsync parameters for object server\n\nChange-Id: I752dc30755febcfa94cb266c2faf3b7aedc35ca9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54c75b9f1f72f927d10edbaab7ddde4934b5e7e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/54c75b9f1f72f927d10edbaab7ddde4934b5e7e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
