)]}'
{"id":"openstack%2Fpuppet-swift~822267","triplet_id":"openstack%2Fpuppet-swift~master~I876c0ab20677976f5ebb54fdd15de0906b6ca323","project":"openstack/puppet-swift","branch":"master","topic":"container_sync-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-23 10:22:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I876c0ab20677976f5ebb54fdd15de0906b6ca323","subject":"Support parameters of container_sync middleware","status":"MERGED","created":"2021-12-19 14:01:18.000000000","updated":"2021-12-23 10:24:29.000000000","submitted":"2021-12-23 10:22:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822267-container_sync-opts","meta_rev_id":"9eaf7246444fcea691543f44bf48c6db33ccfc08","_number":822267,"virtual_id_number":822267,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 10:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-20 15:31:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-12-23 08:32:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-12-23 08:32:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-19 15:36:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:31:49.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-23 08:32:01.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"5289cb2676cf931f75cb135e3dade459c3675683","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:01:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88ee5dc8ade52c83236a3688416ab60867cbc7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 15:36:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a2b9074882447129c403e9a253c3cd1 : FAILURE in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab80cb929ee2461ead2723a359bb5cf1 : FAILURE in 3m 19s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/65c053a65f3346c68ae8d7e0d92dbe23 : FAILURE in 3m 16s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8a049e6c814a4eaa8eae9e72d04dfd27 : FAILURE in 13m 58s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/599c7e4014c0428789a34e8af88eabb9 : FAILURE in 13m 36s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ad49d0cc6f97444e9aa116518827180a : FAILURE in 14m 29s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c434e15a15cb48bd897b5a48ba0c9e26 : FAILURE in 13m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5adeb9daaa94400a7fe02d958654527 : FAILURE in 12m 00s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d346019fb7d3475b8d50cc0997e11f27 : FAILURE in 29m 14s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/e53c4480ce8746f987af88f1998a77ea : SUCCESS in 1h 29m 55s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/7d8242007514422685657190f93eb2bb : SUCCESS in 55m 47s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d65a8e68c9147d2b8014157ac089f9b : FAILURE in 17m 31s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8196980f32124eb488609c2d92703c9d : SUCCESS in 27m 59s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3b47451eb504b83afe6a489aa7e0a57 : FAILURE in 20m 59s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aba8a5d12ba449ddaa8f84c5934db71b : FAILURE in 21m 59s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9704dbb4e56648e5b143790f4327cca4 : SUCCESS in 4m 11s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/fe70ec9eec7f471b914c8226c7b77b72 : FAILURE in 1h 12m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9ca2486cdc75083a7b02055327797559f9761a57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 01:15:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fac83427a0dab3798ba1c7f3d69945476ebee7f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 02:39:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c3d41ab2f454cec8064056081397adf : SUCCESS in 3m 35s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/578dcc64e3e647c39bbd354ef02d771c : SUCCESS in 3m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fdfb025cf3cc4d3c8bfbfeac3925a66c : SUCCESS in 2m 56s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/f734c7423ad846729657abf1b1a91b3c : FAILURE in 15m 07s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f9b586f91c74fac82e493782282ce9f : FAILURE in 15m 08s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db28bc6718bf469eba7dd7ff9009a8bf : FAILURE in 10m 42s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6841c8ef2c2435f877d12a0669e668f : FAILURE in 9m 49s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55fd8453ff1246e0a1ecac9d526d4118 : FAILURE in 12m 43s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a963e7cfa53f47bf9ed3cfb79a67c6a0 : FAILURE in 27m 42s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a97ed573359c448abda23f4d357d73f2 : SUCCESS in 1h 06m 41s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f8301d886d7453f856578978a941df8 : SUCCESS in 48m 15s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54de5e51fbd7421085ba0e1b11aab941 : FAILURE in 19m 17s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0130da4dc3b4559b7f4410bea25304d : SUCCESS in 30m 11s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/663e12e613f84dd3a8c279adc57213dc : FAILURE in 26m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0e5a3caf999b45359ee71809e557ee1b : FAILURE in 19m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dbe1fa0d96634806883835595e016472 : SUCCESS in 3m 17s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/22f178d282044629af78b6a2dc360bf1 : SUCCESS in 1h 21m 40s","accounts_in_message":[],"_revision_number":2},{"id":"1dec9e8bcf4c9c36ec805a0606b9c956dbc0ba08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 07:15:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bf074aea8560758598337c6df4122051b473463e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 07:17:42.000000000","message":"Topic set to container_sync-opts","accounts_in_message":[],"_revision_number":3},{"id":"47f535f7757977f2e7859263390cedf20eac5467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 08:49:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f596682c3dc447bab7728e059df9b929 : SUCCESS in 2m 45s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cabe370f59f4740ab71d1d0dcb9668f : SUCCESS in 3m 18s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/74ea350395c2493a81dd679fdc73a8e2 : SUCCESS in 3m 09s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/c964bd7812e34bccace762cf96ec6817 : SUCCESS in 14m 37s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/24a198a86855478ab6d97cfa683e2df0 : SUCCESS in 14m 27s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5f0b5e941f184f8c98b21b7f3386363e : SUCCESS in 13m 23s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9b3afc066e474668b84074b389dc5f5a : SUCCESS in 13m 00s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/118b0f86fc6848b8b4bdc3b1b01e33a6 : SUCCESS in 12m 54s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8c19c7da1a2b42388bcd62fae0c693f7 : FAILURE in 35m 44s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8ac10e2dae6643319a1998af44dcc0a3 : SUCCESS in 1h 29m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/5026f713282c49d6af3ffceea746537c : SUCCESS in 52m 51s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8e16328558784a5799b0bb8cb6b2cbda : FAILURE in 23m 59s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/b7976198534740e48a33c8e28ff1420a : SUCCESS in 28m 51s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5bece65db404553920a1b09ac032fde : FAILURE in 34m 31s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23af2d45263c4b5d86c6c325f4bd0110 : FAILURE in 18m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a421dfb95e1c44d58cb5de4d5e1f800e : SUCCESS in 3m 52s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/8ae7dd698a6f4cbe9ef046ded852fabf : SUCCESS in 1h 23m 54s","accounts_in_message":[],"_revision_number":3},{"id":"3fd6949231411185d23d32ef6b3353224655220e","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:31:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"556b449afe93a2200b8f396da40c7aafbc6a92c3","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-23 08:32:01.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4f15fcec04f560d75f6ee58c66b163833fa6d5b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 08:33:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e4ed613a4e709d5c3a8b86f4885640fb5575c446","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:22:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/063dc76854bf4f80842e10b6bbd889cf : SUCCESS in 3m 41s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ae13bc4eccc4b4488e1040a62c84337 : SUCCESS in 3m 10s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cd09a2da3ab40f8a99571075e6f2794 : SUCCESS in 3m 57s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/24fd5456413c496d876bea71823e6e8c : SUCCESS in 14m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f802006e5db4719b2759e17e7e3ff64 : SUCCESS in 12m 38s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dfc2a72aedae4a77b1031b1a66f8b3ac : SUCCESS in 12m 21s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1d3e36f26c04cbdb57d3ba44e9623c4 : SUCCESS in 11m 16s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/6f267eca5d3c4e30b1c5d2d5b8ef417f : SUCCESS in 1h 13m 33s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/c781efc18be34d8190ce9ee404de7f94 : SUCCESS in 30m 27s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0310b62350d541fdb6c97669cd4d422e : FAILURE in 22m 03s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2252e223e77c485cabc9bcd9885b527d : SUCCESS in 4m 32s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f5ef2992cd47471a946c31cb936c8fd7 : SUCCESS in 1h 19m 22s","accounts_in_message":[],"_revision_number":3},{"id":"41dcedc3f7f5869d6198ba482ed8e37c6ebdbffa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:22:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9eaf7246444fcea691543f44bf48c6db33ccfc08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:24:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af412306118e419491c83eeb00280513 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a39af5a7dd205e00072d10f140d0f446413936ca","revisions":{"d155b6e4990286422907a4f3fe4214b4c9458bb5":{"kind":"REWORK","_number":1,"created":"2021-12-19 14:01:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/822267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/1"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:00:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:01:10.000000000","tz":540},"subject":"Support parameters of container_sync middleware","message":"Support parameters of container_sync middleware\n\nChange-Id: I876c0ab20677976f5ebb54fdd15de0906b6ca323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d155b6e4990286422907a4f3fe4214b4c9458bb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/d155b6e4990286422907a4f3fe4214b4c9458bb5"}]},"branch":"refs/heads/master"},"1a5966741ed0503fc1d88d763412f384b588e802":{"kind":"REWORK","_number":2,"created":"2021-12-20 01:15:24.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/822267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/2"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:00:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-20 01:15:18.000000000","tz":0},"subject":"Support parameters of container_sync middleware","message":"Support parameters of container_sync middleware\n\nChange-Id: I876c0ab20677976f5ebb54fdd15de0906b6ca323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a5966741ed0503fc1d88d763412f384b588e802"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1a5966741ed0503fc1d88d763412f384b588e802"}]},"branch":"refs/heads/master"},"a39af5a7dd205e00072d10f140d0f446413936ca":{"kind":"REWORK","_number":3,"created":"2021-12-20 07:15:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/67/822267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/67/822267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/67/822267/3"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:00:33.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-20 07:15:22.000000000","tz":540},"subject":"Support parameters of container_sync middleware","message":"Support parameters of container_sync middleware\n\nChange-Id: I876c0ab20677976f5ebb54fdd15de0906b6ca323\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a39af5a7dd205e00072d10f140d0f446413936ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/a39af5a7dd205e00072d10f140d0f446413936ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
