)]}'
{"id":"openstack%2Fpuppet-swift~822269","triplet_id":"openstack%2Fpuppet-swift~master~I29e118af2ad9893ab1cc42a1d6befdfa07395a20","project":"openstack/puppet-swift","branch":"master","topic":"slo-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-23 10:22:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29e118af2ad9893ab1cc42a1d6befdfa07395a20","subject":"Support more [filter:slo] parameters","status":"MERGED","created":"2021-12-19 14:26:30.000000000","updated":"2021-12-23 10:22:39.000000000","submitted":"2021-12-23 10:22:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822269-slo-opts","meta_rev_id":"2c4057cae9b4464e5ab34ffc5e7e347559c0c65e","_number":822269,"virtual_id_number":822269,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 10:22:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-20 15:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-12-23 08:31:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-12-23 08:31:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-19 16:32:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:32:05.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-23 08:31:17.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"c8810d65033edfd377c854647707872dd436d1ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:26:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c184687363c204474ab8d0d9ca1f5cee43b8396","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:56:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"03a8dd098c3c2520dbaf6288c4f4e4189c539237","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 16:32:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89514aeff4bc49769bde5b263a2fadf6 : FAILURE in 3m 20s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0899c4ad956f481a80eea038ba760830 : FAILURE in 4m 01s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/324ae2ccb0e840f881dd06b1820cdd83 : FAILURE in 3m 18s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c5c08a399ff4859870850c4862eeac0 : FAILURE in 14m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/26104112c1c44365889bbb6b3750ef4b : FAILURE in 13m 33s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a471a519bc5245c682ea3d9259ec8c9d : FAILURE in 11m 43s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0340aa5c2284911b80ac8ea8f463ada : FAILURE in 14m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3dcd340a0c24560a075254278b64ee2 : FAILURE in 10m 31s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abe5e1f2500249109cacbe784181958a : FAILURE in 36m 49s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/ad04ad98af714e4ca77b4400e6996e0d : SUCCESS in 1h 33m 21s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/ec394b0c6f4449b0b09008fdd97b4793 : SUCCESS in 45m 01s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b015ae8a1dd848d2906e8a11882dbcd2 : FAILURE in 20m 32s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/164a8516f742491f95253aa361851f4c : SUCCESS in 24m 17s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2168f13752aa435982a0dc251f88ca7d : FAILURE in 20m 58s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e640c69d83b741f087aeae6074c5c94e : FAILURE in 20m 17s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/88455f8092004a0e87ae05b970a6a7e4 : SUCCESS in 3m 25s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/c75566255d2f4bd9b20bbe62c178a367 : FAILURE in 51m 45s","accounts_in_message":[],"_revision_number":2},{"id":"47bb0cd522f0de6bdff626f9f9626e3739933360","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 23:10:15.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e0a5c614b0f737036d25d07978c157b7c2b725a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 00:33:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e47988b8b8f4e6b9ca2fc0e5e0629dd : SUCCESS in 3m 56s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1536b9b2c1214625a01660bc15c356bc : SUCCESS in 5m 39s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8eea04e7cd2402884332c601b0788d6 : SUCCESS in 3m 50s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/46ec9c35705e4e13acee4b6b23e8dc7d : SUCCESS in 14m 38s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b756585b14945a38401a85288f2e457 : SUCCESS in 14m 41s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c482b4546891494a97a0fd4ce8db6dd4 : SUCCESS in 14m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/050926a56c104669a7549c0b4e7de8fc : SUCCESS in 13m 46s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52de5769b44b4fd6a7f8643fe3ab783b : SUCCESS in 13m 06s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4e69aea12542b6961f5426c80c692a : FAILURE in 31m 29s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/c12eab06191541668051cfa72391dc26 : SUCCESS in 1h 18m 40s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/2efd1149d7bf437e9afb98c839900469 : SUCCESS in 1h 09m 24s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a721375f3d3342389b1ab24ac4bb8d48 : FAILURE in 23m 23s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/4dc04ab481a54b25a43098881132e3d6 : SUCCESS in 21m 38s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3b83a6a2ecc4f25b62373fff710bebc : FAILURE in 22m 38s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d25ca2ae101e4153bd976c9b56a35d7b : FAILURE in 24m 16s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3fa192b19e34acb8d29a181e783258b : SUCCESS in 5m 33s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/cb1d2cfc274b4ac98aa51b5a4d811399 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":3},{"id":"b9bbb6701dea335098c8721d3237f7247efeaf4e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 07:17:16.000000000","message":"Topic set to slo-opts","accounts_in_message":[],"_revision_number":3},{"id":"f88e080fd7bc50b4c0a1148b21b9acdb15f53575","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:32:05.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2cb202f5251268b2a60bf6c2d4e42981e75c9a0d","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-23 08:31:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"95ac10bea50b3c45f80e7693a20fc5773e371967","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 08:32:50.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd7d0561c89c4eb004ae5a9f69c53de321d9a58b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:22:39.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c089792090447df8568ec1de0884def : SUCCESS in 3m 42s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1157b483261d435498bae0f3a4901321 : SUCCESS in 3m 49s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/603927b345994f77a7370f5b475bd9e0 : SUCCESS in 4m 17s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/5553602cad45433aaf75d0ceca1c00b3 : SUCCESS in 16m 19s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/643860e07e07445ca888e62ff0ccd333 : SUCCESS in 15m 53s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37b7da5ab52e478a8e23828fdd1b9bdb : SUCCESS in 15m 26s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c7a439eec5b4ef68ca0a1152358f09f : SUCCESS in 15m 42s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/055ac85632cf48e1a06c57b5c016c3a0 : SUCCESS in 1h 28m 02s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/7f5534b750764c6081263d881aafe014 : SUCCESS in 31m 52s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/283a22f5b6da417c8ee35019a5cc2e11 : FAILURE in 26m 30s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ebcf295eede44d4f80c931eff9918617 : SUCCESS in 4m 22s\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/f6dee393f8584a839a170b98016267df : SUCCESS in 1h 22m 14s","accounts_in_message":[],"_revision_number":3},{"id":"2c4057cae9b4464e5ab34ffc5e7e347559c0c65e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:22:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ef9ff551315c09025d43418c80505d18bacaf44b","revisions":{"ed47f1e7ca6973e76219b0e885063c32eb05a6c9":{"kind":"REWORK","_number":1,"created":"2021-12-19 14:26:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/822269/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/69/822269/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/1"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:26:00.000000000","tz":540},"subject":"Support more [filter:slo] parameters","message":"Support more [filter:slo] parameters\n\nChange-Id: I29e118af2ad9893ab1cc42a1d6befdfa07395a20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ed47f1e7ca6973e76219b0e885063c32eb05a6c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ed47f1e7ca6973e76219b0e885063c32eb05a6c9"}]},"branch":"refs/heads/master"},"86a1986e181d6bd904614acf9e9170724c6d01ba":{"kind":"REWORK","_number":2,"created":"2021-12-19 14:56:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/822269/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/69/822269/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/2"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:51:27.000000000","tz":540},"subject":"Support more [filter:slo] parameters","message":"Support more [filter:slo] parameters\n\nChange-Id: I29e118af2ad9893ab1cc42a1d6befdfa07395a20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/86a1986e181d6bd904614acf9e9170724c6d01ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/86a1986e181d6bd904614acf9e9170724c6d01ba"}]},"branch":"refs/heads/master"},"ef9ff551315c09025d43418c80505d18bacaf44b":{"kind":"REWORK","_number":3,"created":"2021-12-19 23:10:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/69/822269/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/69/822269/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/69/822269/3"}}},"commit":{"parents":[{"commit":"1ef0107008411e214f1625577e8a0866fb77a434","subject":"Merge \"Fix ignored unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/1ef0107008411e214f1625577e8a0866fb77a434"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:26:00.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 23:09:35.000000000","tz":540},"subject":"Support more [filter:slo] parameters","message":"Support more [filter:slo] parameters\n\nChange-Id: I29e118af2ad9893ab1cc42a1d6befdfa07395a20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef9ff551315c09025d43418c80505d18bacaf44b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/ef9ff551315c09025d43418c80505d18bacaf44b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
