)]}'
{"id":"openstack%2Fpuppet-swift~822272","triplet_id":"openstack%2Fpuppet-swift~master~Ic140589cf76c8d8f0a82ba259b4940f565cabaa5","project":"openstack/puppet-swift","branch":"master","topic":"slo-opts","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-23 11:14:08.000000000","reason":"Takashi Kajinami replied on the change"}},"hashtags":[],"change_id":"Ic140589cf76c8d8f0a82ba259b4940f565cabaa5","subject":"Do not hard-code defaults of [filter:slo] parameters","status":"MERGED","created":"2021-12-19 14:56:29.000000000","updated":"2021-12-23 15:14:28.000000000","submitted":"2021-12-23 15:14:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"822272-slo-opts","meta_rev_id":"01fb859b696b08b30ad31ceab54737a9c59297db","_number":822272,"virtual_id_number":822272,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-23 15:14:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"all":[{"value":2,"date":"2021-12-20 15:32:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":2,"date":"2021-12-23 08:31:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"all":[{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2021-12-23 08:31:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-19 16:26:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-20 15:32:28.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2021-12-23 08:31:41.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"state":"REVIEWER"}],"messages":[{"id":"7ed48c082fb8c7bb2567962c54e2a845a308e344","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 14:56:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d184f4018327f482cfebdffd1a64ffbc9cf6794e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-19 16:26:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9093801422e4055b3bdbd395adaf0cb : FAILURE in 3m 14s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02260977b744ff78f914e1df376cb09 : FAILURE in 3m 05s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b225041ffb5a45e3b02e7432be01a367 : FAILURE in 4m 00s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/129ab938e8bd475dafcb1729c15d76fa : FAILURE in 12m 39s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/1784aab3e9b74b58a8db2e9a53df907f : FAILURE in 17m 37s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d7afc5012e6417d9bd863eea52bf062 : FAILURE in 10m 49s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fd51afe348844408ff63707017f0f7e : FAILURE in 13m 05s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a2ca9617fac44aea98881d7090c97767 : FAILURE in 13m 11s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55357ca104484764968fb25fd09de3b5 : FAILURE in 31m 09s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/3720added6e94154b3b2017573ec96e2 : SUCCESS in 1h 28m 19s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/cabca1211b624b4a943e97c02bc38f1e : SUCCESS in 50m 40s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/babf5236804b466bbc8bb735710d4aa1 : FAILURE in 20m 02s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/ceb0db9962194856b7aaccf9f929e359 : SUCCESS in 28m 39s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29842a357d324e029a91860828e2e25d : FAILURE in 20m 21s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d948a9d0fd45c4adb05fd826d66945 : FAILURE in 24m 48s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/ab96e48fb38444688b15caf8c6de7b2a : FAILURE in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3dfa63aded12234be0e5012ebd90a2f6cd2c59a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-19 23:10:15.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0455ae63b61c8def771d7ef528ab4ce929cb2852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 00:34:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00d30e10fea14d90ad7e9fb1234db6d2 : SUCCESS in 3m 36s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b578ece2238e4e3d829e974d1e313c2f : SUCCESS in 3m 25s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6489c197cda34aeaa9faa7a53f9a697a : SUCCESS in 3m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/16ae9347ef854d4cae8e486bd1e6f9ee : FAILURE in 15m 15s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/dda85f5b53cf456e996e7992b9a4d560 : FAILURE in 13m 34s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94fbd9b3f0fd437caa81fb8eb3f475c5 : FAILURE in 11m 07s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/737ee8feb9fa4bbfbc336d75aae9f2c1 : FAILURE in 11m 44s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a906036aecba4178b4c2529d32f69559 : FAILURE in 13m 40s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b57ca519799e4cea83e728437f32ac54 : FAILURE in 31m 51s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/a8db73c03ebb4ac6bf29f89aa4889d5a : SUCCESS in 1h 07m 07s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/58e688d7b98b4164af7d2ebfd5bac0fd : SUCCESS in 45m 33s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2cb3cd56d64249dfb78a2b4ab009cc3d : FAILURE in 19m 55s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/6917d85930b94ff09d97bf475aa8e410 : SUCCESS in 26m 09s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/77ed8a60122d45cba4933f8b1a217352 : FAILURE in 25m 34s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7fe1c56fd0043e4ae64420f9bf694a4 : FAILURE in 20m 16s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9086a43f321d4f9e9485d1c057fa19a1 : SUCCESS in 1h 20m 13s","accounts_in_message":[],"_revision_number":2},{"id":"1c26f800fb3053a4f60631a1daa8c1db65295580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 01:10:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3934b81f665f4e6f1ce59e6f1b308b284c539cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 02:39:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82f6801e452e42a7a13311d619beba46 : SUCCESS in 4m 30s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5209b4934623478189a6b2ef62153c4a : SUCCESS in 3m 20s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/452520efed2a436c82d533737ef15ac2 : SUCCESS in 3m 06s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/324fc3c004cb426eb40f51840ddeb3cc : SUCCESS in 16m 16s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/2785e7edee6c4fa6aa58c1ea37e0f042 : SUCCESS in 13m 54s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/52770fbd162f40ea8608c8902ba0ed8b : SUCCESS in 13m 00s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b3e1d88b9d934f879a550cb51a872136 : SUCCESS in 17m 34s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9697ad5839cd4b4da4b506e29371d0c0 : SUCCESS in 11m 30s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eadb52f32c2147349fec965da4c4a913 : FAILURE in 35m 34s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/934acd0679754118aacd69e8f7b04e9f : SUCCESS in 1h 23m 57s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/29423ff8997348ce8e4cb171141358f1 : SUCCESS in 53m 02s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f774a08c77be4626b668edc9a98b9af9 : FAILURE in 22m 20s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/f239508d62a44f4184dd0ed40d980cda : SUCCESS in 27m 36s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e41feb838dd14fcaa3241f3d80bcca22 : FAILURE in 19m 52s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/928697695fb74ac6b5fbf96f30938393 : FAILURE in 21m 54s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/9be283e0b2d147f5a3d86488e6b84235 : SUCCESS in 1h 04m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d0aa70168722c4640f4c52295fe31873eb0230bf","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-20 07:17:25.000000000","message":"Topic set to slo-opts","accounts_in_message":[],"_revision_number":3},{"id":"d9e880c16e0689a5d3dd8eef7f4a26f2c174ec26","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2021-12-20 15:32:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"90d8ad0b8d97f273f35d2a753c6acff6cc993b23","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2021-12-23 08:31:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"04b95f22c1e6412bf0454fdc0ebaa8bcb0f0553a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 08:33:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"43f77e4d9acce6b937614e547bf4d7d62090b4d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 10:22:46.000000000","message":"Patch Set 3: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"d624071f4dc96514e5dae59e3020b5f73cb5c65e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-23 11:14:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ada1301cb79f7b82815547603deee117d59ab0c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 12:51:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3930bc89ae841fe806396c4fdc82e4b : SUCCESS in 3m 32s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a4895a90bbf4ab89e71193a85903802 : SUCCESS in 3m 35s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37b4146c137c444a9e373ca4a2f22c11 : SUCCESS in 3m 20s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/aac2f1ee5dc945c29b523ad885374f67 : SUCCESS in 14m 57s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cd8a78852c7f42f7b57848fad2aae960 : SUCCESS in 15m 03s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ddae7d05bb28454a84e2519a936960a1 : SUCCESS in 13m 55s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/817134c58c784d83a7229b6f2ed1eaf1 : SUCCESS in 14m 12s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5285205cae634db3b10f8d63f6a37fa0 : SUCCESS in 12m 59s (non-voting)\n- puppet-openstack-integration-6-scenario002-tempest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cb283f6e976c41ba852de015f0bfd696 : FAILURE in 37m 25s (non-voting)\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/8692071857f54cbab44fd6c042c5e582 : SUCCESS in 1h 29m 34s\n- puppet-openstack-integration-7-scenario002-tempest-centos-9-stream https://zuul.opendev.org/t/openstack/build/332c77b569914ac6ba3fa1b37d9c209d : SUCCESS in 57m 36s (non-voting)\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42b27869b3e544dc82e5ef9148756546 : FAILURE in 25m 05s (non-voting)\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0e6674cc5c8446f8fcd79820773295b : SUCCESS in 29m 08s\n- puppet-openstack-litmus-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e33c9b0308914041bd59a2d584ffeb65 : FAILURE in 23m 03s (non-voting)\n- puppet-openstack-litmus-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a0f95f95f1a744d9ac5e93e875040c9c : FAILURE in 25m 31s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/99b40eceb7b840c989cafbef8f36b553 : SUCCESS in 1h 25m 35s","accounts_in_message":[],"_revision_number":3},{"id":"9269d6e883b4f352632076bf950a72d7cb19a1f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 13:10:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7acb89d996b7f477142018b6f4c68c0616a6b3fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:14:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7673ee1ca60341628be4bad78c543a76 : SUCCESS in 4m 13s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/baea6f8fc5af48cd99022d837e221600 : SUCCESS in 3m 54s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f7333614f3ee45e4b793b9300a7aee53 : SUCCESS in 4m 28s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/48595fa94f5e43858a4bd311738447e4 : SUCCESS in 15m 03s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbafe0296e9b4a909b1088cfd9fa5249 : SUCCESS in 16m 23s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43188eea3f9d4e25bacd12c1d9e4c9b8 : SUCCESS in 14m 32s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/394feda499b547fa9cb5784b0e976268 : SUCCESS in 15m 59s\n- puppet-openstack-integration-7-scenario002-tempest-centos-8-stream https://zuul.opendev.org/t/openstack/build/30cd5c867010435aa2178d5c3af37def : SUCCESS in 1h 43m 22s\n- puppet-openstack-litmus-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/fa7e2283fb5c4c879c4e9d3e37dc8422 : SUCCESS in 30m 44s\n- puppet-openstack-litmus-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a0f7a41d6254eb0be2eebd0af2224a5 : FAILURE in 24m 09s (non-voting)\n- puppet-swift-tripleo-standalone https://zuul.opendev.org/t/openstack/build/832adfe12f3c4b519c1a38b48573091d : SUCCESS in 1h 17m 49s","accounts_in_message":[],"_revision_number":3},{"id":"01fb859b696b08b30ad31ceab54737a9c59297db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 15:14:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cbf5c60e800438f3781dee0b8563488cad077c23","revisions":{"214b212070cfe15810d4877b8053677b1973e4c1":{"kind":"REWORK","_number":1,"created":"2021-12-19 14:56:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/822272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/72/822272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/1"}}},"commit":{"parents":[{"commit":"4c654d647b21c055388edc82dd88a3c9e348d8db","subject":"Update default value of [filter:slo] max_manifest_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/4c654d647b21c055388edc82dd88a3c9e348d8db"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:48:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:55:19.000000000","tz":540},"subject":"Do not hard-code defaults of [filter:slo] parameters","message":"Do not hard-code defaults of [filter:slo] parameters\n\n... but use the default values defined in swift itself.\n\nChange-Id: Ic140589cf76c8d8f0a82ba259b4940f565cabaa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/214b212070cfe15810d4877b8053677b1973e4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/214b212070cfe15810d4877b8053677b1973e4c1"}]},"branch":"refs/heads/master"},"5c1cc0088c56fab050de0dcca26bf40d4b084da4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-12-19 23:10:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/822272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/72/822272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/2"}}},"commit":{"parents":[{"commit":"7017388604a1d1be50efb7868be26f6c48f065c2","subject":"Update default value of [filter:slo] max_manifest_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7017388604a1d1be50efb7868be26f6c48f065c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:48:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 23:10:00.000000000","tz":540},"subject":"Do not hard-code defaults of [filter:slo] parameters","message":"Do not hard-code defaults of [filter:slo] parameters\n\n... but use the default values defined in swift itself.\n\nChange-Id: Ic140589cf76c8d8f0a82ba259b4940f565cabaa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5c1cc0088c56fab050de0dcca26bf40d4b084da4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5c1cc0088c56fab050de0dcca26bf40d4b084da4"}]},"branch":"refs/heads/master"},"cbf5c60e800438f3781dee0b8563488cad077c23":{"kind":"REWORK","_number":3,"created":"2021-12-20 01:10:23.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/72/822272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/72/822272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/72/822272/3"}}},"commit":{"parents":[{"commit":"7017388604a1d1be50efb7868be26f6c48f065c2","subject":"Update default value of [filter:slo] max_manifest_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/7017388604a1d1be50efb7868be26f6c48f065c2"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-19 14:48:58.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-20 01:10:12.000000000","tz":540},"subject":"Do not hard-code defaults of [filter:slo] parameters","message":"Do not hard-code defaults of [filter:slo] parameters\n\n... but use the default values defined in swift itself.\n\nChange-Id: Ic140589cf76c8d8f0a82ba259b4940f565cabaa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cbf5c60e800438f3781dee0b8563488cad077c23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/cbf5c60e800438f3781dee0b8563488cad077c23"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
