)]}'
{"id":"openstack%2Fpuppet-swift~822773","triplet_id":"openstack%2Fpuppet-swift~master~I292931305278414eadb368a8b4931d200a12d2e1","project":"openstack/puppet-swift","branch":"master","topic":"c9s-ut","hashtags":[],"change_id":"I292931305278414eadb368a8b4931d200a12d2e1","subject":"DNM: Test unit tests for CentOS9 Stream","status":"ABANDONED","created":"2021-12-23 02:24:02.000000000","updated":"2021-12-23 08:30:07.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"15880485b8a0da3f1b7b61915d219348ce3244b2","_number":822773,"virtual_id_number":822773,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-23 03:06:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-23 03:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e913334cd840822c430dea986dd71b3aef696cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-23 02:24:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43f4491499103e39a36d43b6e6de08a172c463e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-23 03:06:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- puppet-openstack-lint-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a007c7599f04450087f84c542726419c : SUCCESS in 5m 02s\n- puppet-openstack-syntax-6-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b28724eed73a47d981c6fb35cf2ed31e : SUCCESS in 6m 16s\n- puppet-openstack-syntax-7-ubuntu-focal https://zuul.opendev.org/t/openstack/build/79552b2ffd32468abf53da36ae2c68cf : SUCCESS in 5m 23s\n- puppet-openstack-unit-6.21-centos-8-stream https://zuul.opendev.org/t/openstack/build/a38bc8ad0c064fc488f19ee5af288e5e : SUCCESS in 19m 45s\n- puppet-openstack-unit-7.8-centos-8-stream https://zuul.opendev.org/t/openstack/build/756a44a4ec3040a8a1b7b73acacdcdb0 : SUCCESS in 18m 32s\n- puppet-openstack-unit-6.21-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e3428bb6f9774d8697343479b85f267b : SUCCESS in 19m 01s\n- puppet-openstack-unit-7.8-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e5a17319f44943e78187a7e7c60badfe : SUCCESS in 18m 02s\n- puppet-openstack-unit-latest-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3577adf765e845d1b737cd2a664e04d4 : SUCCESS in 22m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"15880485b8a0da3f1b7b61915d219348ce3244b2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-23 08:30:07.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f4447fd8898b926953f1637f76ff033d699b8ecc","revisions":{"f4447fd8898b926953f1637f76ff033d699b8ecc":{"kind":"REWORK","_number":1,"created":"2021-12-23 02:24:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/73/822773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/puppet-swift","ref":"refs/changes/73/822773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/822773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/822773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/puppet-swift refs/changes/73/822773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/puppet-swift refs/changes/73/822773/1"}}},"commit":{"parents":[{"commit":"5a7a299a544809eca94ff59a712caffd4184553e","subject":"CentOS9: Fix unit tests related to use_xinetd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/5a7a299a544809eca94ff59a712caffd4184553e"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-23 02:23:52.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-12-23 02:23:52.000000000","tz":540},"subject":"DNM: Test unit tests for CentOS9 Stream","message":"DNM: Test unit tests for CentOS9 Stream\n\nDepends-on: https://review.opendev.org/822769\nChange-Id: I292931305278414eadb368a8b4931d200a12d2e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f4447fd8898b926953f1637f76ff033d699b8ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/puppet-swift/commit/f4447fd8898b926953f1637f76ff033d699b8ecc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
